)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~735979","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I29f201332286eb975e522802c6d96cc892ad2cd2","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"zuul-build-workspace-uuid","hashtags":[],"change_id":"I29f201332286eb975e522802c6d96cc892ad2cd2","subject":"WIP: Update \u0027src/\u0027 by prepending zuul build uuid","status":"ABANDONED","created":"2020-06-16 17:35:19.000000000","updated":"2022-07-05 08:30:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"cacf043b23635d699c698cada9ea1cca5f811bcb","_number":735979,"virtual_id_number":735979,"owner":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-16 22:12:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 22:12:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d94942f0f671492dc793a22530e96db458bdf0f","author":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"date":"2020-06-16 17:35:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ab8f4bd908b84a2fd2dad64bb6825cdb3da342f","author":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"date":"2020-06-16 21:55:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"df1acdd17373b8a209062e498457a2930346572d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 22:12:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b46a9e015268425a84838ac2a9a0b37b : SUCCESS in 6m 07s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/25c9e1b65fc44af680c79f87a3f45ae4 : SUCCESS in 5m 04s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/ffb191896bfe4e86857ad0702d313942 : SUCCESS in 4m 21s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6cf2aa1d8e7b4b0ab504d87769317353 : FAILURE in 7m 46s\n- build-wheel-cache-centos-7 https://zuul.opendev.org/t/openstack/build/b6df2c637a644a7db3aacf1c8f31560e : POST_FAILURE in 5m 45s\n- build-wheel-cache-centos-8 https://zuul.opendev.org/t/openstack/build/c108d9d7999e48a8a08ae94875908747 : POST_FAILURE in 5m 53s\n- build-wheel-cache-debian-buster https://zuul.opendev.org/t/openstack/build/78acf24ed6b34c3a8831d7612563168e : POST_FAILURE in 4m 53s\n- build-wheel-cache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/64c9c4a0d80043bc94f26d1321a40e3e : POST_FAILURE in 5m 27s\n- build-wheel-cache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7d6c6893f6f469b9c23d4993ee3af3c : POST_FAILURE in 5m 06s\n- build-wheel-cache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1464ebad19c948dc8fbde45cc0666ee6 : POST_FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"95dd8cfff1ae7a50a2df57734b0585f5c6466611","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:22:30.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- build-wheel-cache-centos-8-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- build-wheel-cache-debian-buster-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- build-wheel-cache-ubuntu-xenial-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- build-wheel-cache-ubuntu-bionic-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- build-wheel-cache-ubuntu-focal-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"07d4bce3499116f0454ec13bd1abe61fe892d3f8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"date":"2020-11-22 10:59:52.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"cacf043b23635d699c698cada9ea1cca5f811bcb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"date":"2022-07-05 08:30:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5cf606793f4d8e7c7989afb2bd94f51551edc409","revisions":{"b0a0eaf39c5baec66883b5ccc0a8f31e600b6088":{"kind":"REWORK","_number":1,"created":"2020-06-16 17:35:19.000000000","uploader":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"ref":"refs/changes/79/735979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/79/735979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/1"}}},"commit":{"parents":[{"commit":"ece7afc50fdf6241dad5e28f29f462e3ebf0bd36","subject":"Merge \"Fix pdf building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ece7afc50fdf6241dad5e28f29f462e3ebf0bd36"}]}],"author":{"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","date":"2020-06-16 14:37:21.000000000","tz":120},"committer":{"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","date":"2020-06-16 14:53:13.000000000","tz":120},"subject":"WIP: Update legacy pre.yaml to use build uuid in repo_dir_src ?","message":"WIP: Update legacy pre.yaml to use build uuid in repo_dir_src ?\n\nWhen using max-parallel-jobs\u003e1 for nodepool static provider, 2+ jobs will\noverlap and prepare-workspace will remove files created by an already running\njobs\nThus \u0027uuid\u0027 is added to \u0027src_dir\u0027\n\nDepends-On: Circular Dependency Ib78ce4272995680f37ddede73fe6622bcb5c9443\nChange-Id: I29f201332286eb975e522802c6d96cc892ad2cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b0a0eaf39c5baec66883b5ccc0a8f31e600b6088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b0a0eaf39c5baec66883b5ccc0a8f31e600b6088"}]},"branch":"refs/heads/master"},"5cf606793f4d8e7c7989afb2bd94f51551edc409":{"kind":"REWORK","_number":2,"created":"2020-06-16 21:55:36.000000000","uploader":{"_account_id":31905,"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","username":"gchauvel"},"ref":"refs/changes/79/735979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/79/735979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/735979/2"}}},"commit":{"parents":[{"commit":"ece7afc50fdf6241dad5e28f29f462e3ebf0bd36","subject":"Merge \"Fix pdf building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ece7afc50fdf6241dad5e28f29f462e3ebf0bd36"}]}],"author":{"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","date":"2020-06-16 14:37:21.000000000","tz":120},"committer":{"name":"Guillaume Chauvel","email":"guillaume.chauvel@gmail.com","date":"2020-06-16 21:49:59.000000000","tz":120},"subject":"WIP: Update \u0027src/\u0027 by prepending zuul build uuid","message":"WIP: Update \u0027src/\u0027 by prepending zuul build uuid\n\nWhen using max-parallel-jobs\u003e1 for nodepool static provider, 2+ jobs will\noverlap and prepare-workspace will remove files created by an already running\njobs\nThus \u0027src/\u0027 is prepended with zuul build \u0027uuid\u0027\n\nDepends-On: Circular Dependency Ib78ce4272995680f37ddede73fe6622bcb5c9443\nChange-Id: I29f201332286eb975e522802c6d96cc892ad2cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5cf606793f4d8e7c7989afb2bd94f51551edc409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5cf606793f4d8e7c7989afb2bd94f51551edc409"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
