)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~736786","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I3947088cec276e80c26591de21efc8013b54d2d4","project":"openstack/openstack-zuul-jobs","branch":"master","hashtags":[],"change_id":"I3947088cec276e80c26591de21efc8013b54d2d4","subject":"(fix) Changed pip to pip3","status":"ABANDONED","created":"2020-06-18 17:29:42.000000000","updated":"2020-06-23 19:52:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"923c3e6e2bed6f04735769fb3cbc8fd34de7084c","_number":736786,"virtual_id_number":736786,"owner":{"_account_id":28664,"name":"Ahmad Mahmoudi","email":"ahmad.mahmoudi@att.com","username":"ahmad"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-18 17:50:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-06-19 06:29:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 17:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-19 06:29:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"741acf564d983a0b44cb6815f043fa02d19bf146","author":{"_account_id":28664,"name":"Ahmad Mahmoudi","email":"ahmad.mahmoudi@att.com","username":"ahmad"},"date":"2020-06-18 17:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ada9ca31e96f8e2643160a70048dea213044de58","author":{"_account_id":28664,"name":"Ahmad Mahmoudi","email":"ahmad.mahmoudi@att.com","username":"ahmad"},"date":"2020-06-18 17:38:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98ed9e6c2f06801f655267c13f5eca5063bd074d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 17:50:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/67ca874d101d47599d930cd50d419420 : SUCCESS in 5m 18s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/dcd665d66cfe4a438936c7245bcb5275 : SUCCESS in 2m 24s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/08754de4df844ec2a456085254e7a678 : SUCCESS in 3m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dac4bf0a17104d3380eb459ee2ee8cf4 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"68cfbd58c0f18c086120c5541a5498dfbd9460b3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-18 17:57:48.000000000","message":"Patch Set 2:\n\nWhich problems are you fixing? Please point to log files","accounts_in_message":[],"_revision_number":2},{"id":"c32fca16341fe83209f576e01592e6d6ba9a9bf3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-19 06:29:29.000000000","message":"Patch Set 2: Code-Review-1\n\nWhy is this needed?","accounts_in_message":[],"_revision_number":2},{"id":"923c3e6e2bed6f04735769fb3cbc8fd34de7084c","author":{"_account_id":28664,"name":"Ahmad Mahmoudi","email":"ahmad.mahmoudi@att.com","username":"ahmad"},"date":"2020-06-23 19:52:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e0821c9bd8694d858d0335893a11e2484c9e846","revisions":{"acbfbfdbba89ac0ca32365b07702926087b1fff9":{"kind":"REWORK","_number":1,"created":"2020-06-18 17:29:42.000000000","uploader":{"_account_id":28664,"name":"Ahmad Mahmoudi","email":"ahmad.mahmoudi@att.com","username":"ahmad"},"ref":"refs/changes/86/736786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/86/736786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/1"}}},"commit":{"parents":[{"commit":"18b046c7b3933ceaa5650ea4a032790905f51f84","subject":"Merge \"Revert \"Work around uwsgi breakage\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/18b046c7b3933ceaa5650ea4a032790905f51f84"}]}],"author":{"name":"Mahmoudi, Ahmad (am495p)","email":"ahmad.mahmoudi@att.com","date":"2020-06-18 17:29:31.000000000","tz":0},"committer":{"name":"Mahmoudi, Ahmad (am495p)","email":"ahmad.mahmoudi@att.com","date":"2020-06-18 17:29:42.000000000","tz":0},"subject":"(fix) Changed pip to pip3","message":"(fix) Changed pip to pip3\n\nChanged pip to pip3 for zuul gates.\n\nChange-Id: I3947088cec276e80c26591de21efc8013b54d2d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/acbfbfdbba89ac0ca32365b07702926087b1fff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/acbfbfdbba89ac0ca32365b07702926087b1fff9"}]},"branch":"refs/heads/master"},"8e0821c9bd8694d858d0335893a11e2484c9e846":{"kind":"REWORK","_number":2,"created":"2020-06-18 17:38:22.000000000","uploader":{"_account_id":28664,"name":"Ahmad Mahmoudi","email":"ahmad.mahmoudi@att.com","username":"ahmad"},"ref":"refs/changes/86/736786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/86/736786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/86/736786/2"}}},"commit":{"parents":[{"commit":"18b046c7b3933ceaa5650ea4a032790905f51f84","subject":"Merge \"Revert \"Work around uwsgi breakage\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/18b046c7b3933ceaa5650ea4a032790905f51f84"}]}],"author":{"name":"Mahmoudi, Ahmad (am495p)","email":"ahmad.mahmoudi@att.com","date":"2020-06-18 17:29:31.000000000","tz":0},"committer":{"name":"Mahmoudi, Ahmad (am495p)","email":"ahmad.mahmoudi@att.com","date":"2020-06-18 17:38:16.000000000","tz":0},"subject":"(fix) Changed pip to pip3","message":"(fix) Changed pip to pip3\n\nChanged pip to pip3 for zuul gates.\n\nChange-Id: I3947088cec276e80c26591de21efc8013b54d2d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8e0821c9bd8694d858d0335893a11e2484c9e846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8e0821c9bd8694d858d0335893a11e2484c9e846"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
