)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~737367","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I181819a614816cba252eafe0f6ad2655b5edfa9a","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"ensure-tox","hashtags":[],"change_id":"I181819a614816cba252eafe0f6ad2655b5edfa9a","subject":"Add ensure-tox to javascript pre roles","status":"ABANDONED","created":"2020-06-22 19:43:29.000000000","updated":"2020-06-23 07:03:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0f9c80058e0b438aa32d096a2cba60c6563bb33","_number":737367,"virtual_id_number":737367,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"date":"2020-06-22 20:05:05.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-22 20:00:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":-1,"date":"2020-06-22 20:18:58.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-22 19:43:39.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-06-22 19:43:43.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-06-22 20:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 20:05:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"72c12a1428739f86a7b439ecc9c8e32dac187b5b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 19:43:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8897f69fe255881fba81674ca19a3a4b897aba41","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 19:51:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2fd69a736c44fbf442ed8446ef587109bffc8594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 20:00:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ea82415c17344adeb39a36ac6d67303f : SUCCESS in 5m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d309793224e04703ac8aa8153a57a947 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e1b753857660f5e9deea12a6294b90010eaf56f4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-22 20:02:57.000000000","message":"Patch Set 2:\n\nWhat is failing? horizon npm scripts work fine after the fix that I did recently...\n\nCould you point to a recent failure, please?","accounts_in_message":[],"_revision_number":2},{"id":"d5b66980a354dbac81e1216056442b1e08e48c07","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-22 20:05:05.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nLooking at your reference, I expect this to be fixed already in horizon master - and a backport of that is needed for stable branches.\n\nbut please give details of a failure to be sure.","accounts_in_message":[],"_revision_number":2},{"id":"cf1a8ab52bf594fbabccbe37d3c6513ea7ca7924","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 20:13:03.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Looking at your reference, I expect this to be fixed already in\n \u003e horizon master - and a backport of that is needed for stable\n \u003e branches.\n \u003e \n \u003e but please give details of a failure to be sure.\n\nLooks like the issue is that change was only to a couple of child jobs, so anything else using the base job still is missing the role.\n\nThis was seen in a patch that is trying to add an ironic cross job to requirements:\n\nhttps://zuul.opendev.org/t/openstack/build/91519dd380934013bdc8ed0cffc1cfb9\n\nI\u0027m still trying to wrap my head around the multiple layers of inheritance coming from different places, so it\u0027s possible I don\u0027t have it in the right spot either.","accounts_in_message":[],"_revision_number":2},{"id":"18bdb0ec41fe108fde6e7c65175c13893cc35d90","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 20:13:40.000000000","message":"Patch Set 2:\n\nIt does look like this fixes the issue though:\n\nhttps://369f8c51f488988cf44a-3bb73954319247ffa3799ea16240aaf1.ssl.cf1.rackcdn.com/734463/5/check/cross-ironic-py38/8200f1b/","accounts_in_message":[],"_revision_number":2},{"id":"56dd1512dfe8f1796fed83b567c9291165918ed8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-22 20:18:58.000000000","message":"Patch Set 2: Workflow-1\n\n\u003e It does look like this fixes the issue though:\n \u003e \n \u003e https://369f8c51f488988cf44a-3bb73954319247ffa3799ea16240aaf1.ssl.cf1.rackcdn.com/734463/5/check/cross-ironic-py38/8200f1b/\n\nSorry, wrong link. It does appear this is still failing, so this isn\u0027t the correct fix either.","accounts_in_message":[],"_revision_number":2},{"id":"1e864a49fb7f158ab8da72d501b85093c4a419e7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-23 07:03:45.000000000","message":"Abandoned\n\nFixed by https://review.opendev.org/#/c/737351/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53a66b907eb0c897b86f739526a4b51a9a6e9f59","revisions":{"e22ec3fadae0476662ad359c6c33fd6790e3591a":{"kind":"REWORK","_number":1,"created":"2020-06-22 19:43:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/737367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/67/737367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/1"}}},"commit":{"parents":[{"commit":"27897ee9deafac6a8a7f1bc8c0c917abdbcb3155","subject":"Symlink tox into the exec path for release jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/27897ee9deafac6a8a7f1bc8c0c917abdbcb3155"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 19:41:40.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 19:41:40.000000000","tz":-300},"subject":"Add ensure-tox to javascript pre roles","message":"Add ensure-tox to javascript pre roles\n\nHorizon and other UI projects call tox in their npm postinstall scripts.\nThis adds the ensure-tox role to the javascript playbook to make sure\nthis script doesn\u0027t error out by not finding tox.\n\nhttps://opendev.org/openstack/horizon/src/branch/master/package.json#L22\n\nChange-Id: I181819a614816cba252eafe0f6ad2655b5edfa9a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/e22ec3fadae0476662ad359c6c33fd6790e3591a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/e22ec3fadae0476662ad359c6c33fd6790e3591a"}]},"branch":"refs/heads/master"},"53a66b907eb0c897b86f739526a4b51a9a6e9f59":{"kind":"REWORK","_number":2,"created":"2020-06-22 19:51:17.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/67/737367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/67/737367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/67/737367/2"}}},"commit":{"parents":[{"commit":"27897ee9deafac6a8a7f1bc8c0c917abdbcb3155","subject":"Symlink tox into the exec path for release jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/27897ee9deafac6a8a7f1bc8c0c917abdbcb3155"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 19:41:40.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-22 19:51:11.000000000","tz":-300},"subject":"Add ensure-tox to javascript pre roles","message":"Add ensure-tox to javascript pre roles\n\nHorizon and other UI projects call tox in their npm postinstall scripts.\nThis adds the ensure-tox role to the javascript playbook to make sure\nthis script doesn\u0027t error out by not finding tox.\n\nhttps://opendev.org/openstack/horizon/src/branch/master/package.json#L22\n\nChange-Id: I181819a614816cba252eafe0f6ad2655b5edfa9a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/53a66b907eb0c897b86f739526a4b51a9a6e9f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/53a66b907eb0c897b86f739526a4b51a9a6e9f59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
