)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~751985","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I61ed9d13f39e88a433d0db2feb5225a1bec861d2","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"fix-xenial-docs-job","hashtags":[],"change_id":"I61ed9d13f39e88a433d0db2feb5225a1bec861d2","subject":"openstack-tox-docs: Enable PDF build for stein-ussuri xenial job","status":"MERGED","created":"2020-09-15 08:13:51.000000000","updated":"2020-09-15 09:39:41.000000000","submitted":"2020-09-15 09:34:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751985-1600162494080-b51fe808","meta_rev_id":"cd3af467d3bf79cf4d9d7de3c2b11d9c69fc3942","_number":751985,"virtual_id_number":751985,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-09-15 09:22:16.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-15 09:34:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-09-15 09:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-09-15 09:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-15 09:25:28.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-15 09:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6fc5e8c304ce9ad5266770f518f0de6c3d3003e6","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-15 08:13:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2829f0f06fc175bd3fb67f8b30808ac43577f37a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-15 08:19:00.000000000","message":"Patch Set 1:\n\nBeing tested by https://review.opendev.org/#/c/751026/4","accounts_in_message":[],"_revision_number":1},{"id":"011ef514fee40b0afa20935f2849caa543eccda5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:21:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fdb2113884714a4ebbb112df8b05510c : SUCCESS in 3m 56s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/4cfc753101b244ed88fd45cb4c359e23 : SUCCESS in 2m 54s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/8d60a04cc05f44b78d0f3187d6d0a288 : SUCCESS in 4m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4e957bc5f7a0464c85e2d1cbb712d171 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"cb34073f8f714d2cc940cd481d227e15bd032887","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-15 08:34:24.000000000","message":"Patch Set 1:\n\nLGTM, waiting for test change to finish.","accounts_in_message":[],"_revision_number":1},{"id":"a739eb80828765fabff36a88516305a07cfc578b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-15 09:22:16.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/#/c/751026/4 confirmed the fix works expectedly.","accounts_in_message":[],"_revision_number":1},{"id":"76b939355449c89ea54c7d7969284a7beedac6a8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-15 09:25:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSingle core approving to unbreak the branches as this was missed in the original change.","accounts_in_message":[],"_revision_number":1},{"id":"01d54ab8062d3e3f5558f056e9e2b27ef752c50f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c12348c07205ef9ffdb615bb37c377b479863c4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:34:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c4a3c910096743b0b4ac46cf655e1274 : SUCCESS in 3m 10s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eb9226ad648244729dfb513e166952ea : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b8ecec2592c0f12644b37ac63f851be9ed413f87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:34:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6546f80f6d1dfdd330b9466da38bd0c13eb369b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 09:39:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/c8b6887ba7ef49f7be5c653aceea21d9 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"785f4c9b2752b678a6522506535e51a5f0d10534","revisions":{"785f4c9b2752b678a6522506535e51a5f0d10534":{"kind":"REWORK","_number":1,"created":"2020-09-15 08:13:51.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/85/751985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/85/751985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/85/751985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/85/751985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/85/751985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/85/751985/1"}}},"commit":{"parents":[{"commit":"c5f427b006a3efbe4a4a7505e7561fce60393a7d","subject":"Remove ubuntu-bionic nodeset from openstack-tox-py27","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c5f427b006a3efbe4a4a7505e7561fce60393a7d"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-15 08:07:39.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-09-15 08:07:39.000000000","tz":540},"subject":"openstack-tox-docs: Enable PDF build for stein-ussuri xenial job","message":"openstack-tox-docs: Enable PDF build for stein-ussuri xenial job\n\nhttps://review.opendev.org/#/c/738326/ created openstack-tox-docs\njob for stein-ussuri branches, but it did not specify pre-run and\nrun phases required for PDF build.\nRepositories which enable PDF build with RSVG support requires\nlibrsvg2-bin even for HTNL build (as rsvg-converter is always\nspecified in doc/source/conf.py). As a result, the above change\nbreaks HTML doc build in stein-ussuri branches for such repos.\nAn example is found at [1] (networking-ovn stable/train branch).\nThis commit makes openstack-tox-docs stein-ussuri same as\nthat for the master branch.\n\n[1] https://zuul.opendev.org/t/openstack/build/0cd86df4b6aa4896ba033d68d4357948\n\nChange-Id: I61ed9d13f39e88a433d0db2feb5225a1bec861d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/785f4c9b2752b678a6522506535e51a5f0d10534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/785f4c9b2752b678a6522506535e51a5f0d10534"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
