)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~753545","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I9dd155227b1f4d35da8cbb09e99141758360d9ed","project":"openstack/openstack-zuul-jobs","branch":"master","hashtags":[],"change_id":"I9dd155227b1f4d35da8cbb09e99141758360d9ed","subject":"Limit monasca-persister legacy jobs to stable","status":"ABANDONED","created":"2020-09-23 08:42:34.000000000","updated":"2020-10-29 10:46:37.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e2327959e03e1376d327e4b3df97d7c7ce5ef1f","_number":753545,"virtual_id_number":753545,"owner":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-10-29 10:35:41.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-23 08:51:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-1,"date":"2020-09-28 19:01:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2020-09-28 19:19:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-29 10:44:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 08:51:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 19:01:58.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-28 19:19:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-10-29 10:35:41.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-10-29 10:44:06.000000000","updated_by":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"reviewer":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"state":"REVIEWER"}],"messages":[{"id":"4f189c69387a5f7afac48c5b2579baedce5abe50","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-09-23 08:42:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19e97f0d3622ca25314e600ea16efaef0203409e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:42:57.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/openstack-zuul-jobs/zuul.d/zuul-legacy-jobs.yaml@master\", line 263, column 23","accounts_in_message":[],"_revision_number":1},{"id":"3d01cad70347b1f7b453fd1dc4e8bd880a068416","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:43:04.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/openstack-zuul-jobs/zuul.d/zuul-legacy-jobs.yaml@master\", line 263, column 23","accounts_in_message":[],"_revision_number":1},{"id":"f525791d271f91986621fcf66e3ab5d97d02120d","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-09-23 08:44:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b868c8381e944ca5305a22f4ff72b6fff716f18a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 08:51:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c839f594e34844748290258aae3449f3 : SUCCESS in 4m 20s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/151d4d22d02b47a186e553fd3b97a6cf : SUCCESS in 2m 51s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/91f928d1ae80468a93a8f3b86d55b6cc : SUCCESS in 2m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7b17e7a17d9b4c278e10f6e2ce87e6b4 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":2},{"id":"07bcff85f940f0419bfa8c5e96fa5f1ea705db05","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-23 18:09:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"406b529b2952abd332a20f55de02e567f8f5a593","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-09-28 12:28:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ed1d1aa603a44d1c81f77dc6abacb73047d98a33","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-28 19:01:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f1f89c53b026ef44bd2a1c3288d5ae2e1f578b9c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-28 19:19:55.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"29106439d2cb52befae01644ed1786e78b0950bf","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-10-29 10:35:41.000000000","message":"Patch Set 2:\n\nI guess this can be abandoned, the following patches have been merged in monasca-persister:\n- https://review.opendev.org/754920\n- https://review.opendev.org/758409 (victoria backport)","accounts_in_message":[],"_revision_number":2},{"id":"b19be39b353b9c69e0219f857cb5ef0defde0afd","author":{"_account_id":28062,"name":"Martin Chacon Piza","email":"martin@chaconpiza.com","username":"martinchacon"},"date":"2020-10-29 10:44:06.000000000","message":"Patch Set 2: -Code-Review\n\nI agree, it can be abandoned.","accounts_in_message":[],"_revision_number":2},{"id":"5e2327959e03e1376d327e4b3df97d7c7ce5ef1f","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-10-29 10:46:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6bef58401096b453a0edeb8c1ffce04ec859202a","revisions":{"4f5edb0eb8548ccbcfee74a4a642baca616884ff":{"kind":"REWORK","_number":1,"created":"2020-09-23 08:42:34.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/45/753545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/45/753545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/1"}}},"commit":{"parents":[{"commit":"7a2f7ec0304e1ffb0ba71f4d97befe03501d574a","subject":"Merge \"Replace texlive-full with cm-super\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/7a2f7ec0304e1ffb0ba71f4d97befe03501d574a"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-09-23 08:39:58.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-09-23 08:42:32.000000000","tz":120},"subject":"Limit monasca-persister legacy jobs to stable","message":"Limit monasca-persister legacy jobs to stable\n\nmonasca-persister Java implementation has been deprecated. We stop\nbuilding and publishing Java binary with Ussuri and do it only on stable\nbranches.\n\nChange-Id: I9dd155227b1f4d35da8cbb09e99141758360d9ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/4f5edb0eb8548ccbcfee74a4a642baca616884ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/4f5edb0eb8548ccbcfee74a4a642baca616884ff"}]},"branch":"refs/heads/master"},"6bef58401096b453a0edeb8c1ffce04ec859202a":{"kind":"REWORK","_number":2,"created":"2020-09-23 08:44:26.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/45/753545/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/45/753545/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/45/753545/2"}}},"commit":{"parents":[{"commit":"7a2f7ec0304e1ffb0ba71f4d97befe03501d574a","subject":"Merge \"Replace texlive-full with cm-super\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/7a2f7ec0304e1ffb0ba71f4d97befe03501d574a"}]}],"author":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-09-23 08:39:58.000000000","tz":120},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-09-23 08:44:17.000000000","tz":120},"subject":"Limit monasca-persister legacy jobs to stable","message":"Limit monasca-persister legacy jobs to stable\n\nmonasca-persister Java implementation has been deprecated. We stop\nbuilding and publishing Java binary with Ussuri and do it only on stable\nbranches.\n\nChange-Id: I9dd155227b1f4d35da8cbb09e99141758360d9ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6bef58401096b453a0edeb8c1ffce04ec859202a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6bef58401096b453a0edeb8c1ffce04ec859202a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
