)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~757101","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I738b833109e4caeb58bb391d79d6e63284462bd8","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"fix/nodejs_doc","hashtags":[],"change_id":"I738b833109e4caeb58bb391d79d6e63284462bd8","subject":"Create nodejs10-doc project templates","status":"MERGED","created":"2020-10-09 13:38:36.000000000","updated":"2021-02-10 04:58:08.000000000","submitted":"2021-02-10 04:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"757101-fix/nodejs_doc","meta_rev_id":"8e74f3dc56f44343bc82d0520542fc42aa11e4b5","_number":757101,"virtual_id_number":757101,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-10 04:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-10-15 08:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-10-25 19:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2021-02-03 18:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-10 00:38:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2021-02-10 04:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"date":"2020-10-29 10:08:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-10-29 10:09:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-10-29 10:09:08.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2021-02-10 04:46:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-13 17:22:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-10-15 08:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-15 08:18:35.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-10-25 19:16:58.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-10-29 10:08:27.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-10-29 10:08:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-10-29 10:09:08.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-10-29 10:09:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2021-01-26 10:20:41.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-02-03 18:40:56.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-02-04 13:00:14.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REMOVED"}],"messages":[{"id":"6a828fd56471f413bbaaf32daee08dd44110a3f7","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-10-09 13:38:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a032c7645399c267313dc72d72543f3c34dda596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 13:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/556cd73508814acba3503effe9fa1f74 : SUCCESS in 2m 49s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/c0363ec467cf4fa1875dc646d23ac489 : SUCCESS in 2m 29s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/0bc63705257a47e3b2805cbedd34fbc2 : SUCCESS in 2m 33s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/14772b74562447d19d054ddd079b0842 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9e8d29875f67180427c2825b80107809401cb53f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-13 11:44:43.000000000","message":"Patch Set 1:\n\nDid you test that these jobs work without changes? I don\u0027t want to break old releases...\n\nPlease push a test change that depends-on this one and add a link here.","accounts_in_message":[],"_revision_number":1},{"id":"7a9d881220170d2075bd9d4bd414e4225016d59a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-10-13 17:22:53.000000000","message":"Patch Set 1:\n\nlet\u0027s add nodejs-run-docs for nodejs 10 version and keep old one as it is. and that we will test it in https://review.opendev.org/#/c/740373/8 and its base","accounts_in_message":[],"_revision_number":1},{"id":"8d19b595def47e22ec8bd4d33cbe8224d74e86e1","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-10-15 07:11:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21db50f99f8ea5212223222c86614617f4e177d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 07:22:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/926fb7afea934a2096508d3756734772 : SUCCESS in 3m 30s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/d61b249166704a9380ac5019f1405ba3 : SUCCESS in 2m 53s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/01ad83cc6f7440aaa82eddbe9cd998d0 : SUCCESS in 2m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e17605913d4b409685da7af566fb2bb8 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c2b55bbbdef8296929ea867cc01688c11e2694a2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 07:45:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f25765c90144d080e1ca2be4aa1973f7ac32d646","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 07:45:50.000000000","message":"Patch Set 2: Code-Review-1\n\nCould you update the commit message so that it reflects your change, please?\n\nchange is fine, just does not match what the commit message says.","accounts_in_message":[],"_revision_number":2},{"id":"d08a494e21f43902c27b45b9de8ebe770f0b60f2","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2020-10-15 07:59:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ec2cd3b6db56aac2d7043f5435a8989dec7df259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 08:13:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ac4fd8635e0048c08934bc3d296074fa : SUCCESS in 3m 58s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/7045068324414891ae35eaa51aea257f : SUCCESS in 2m 54s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/b010a9505e59437f823c0824488fef50 : SUCCESS in 3m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5602e05338cf45be9b86c51bbc865fc5 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"c735b7fa27f43325fc751c3f57cb36134ce27a4f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-15 08:18:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"354128e920456b05d4be965c8ed3bfdc4a78f55c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-10-25 19:16:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8d1b1d23209f679abcd43a82994fc09b645b50c4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-11-25 13:23:54.000000000","message":"Patch Set 3:\n\nrecheck to get fresh logs and bump this in the review list","accounts_in_message":[],"_revision_number":3},{"id":"b1f2481d155a28bc61d4db4d3436fd98f99500f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 13:45:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a314b79a0fbe405ebc51b9b04325b422 : SUCCESS in 4m 40s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/5f376342db0749068856a4f80b8c7407 : SUCCESS in 2m 52s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/cb2bece5ac5543e3a9aae2c69828fb10 : SUCCESS in 4m 49s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3792bdec01e049089cb9a7e4829d4b47 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"3ce6494aff65ca599bdd8b4727b6a356b5ce170f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-26 10:34:14.000000000","message":"Patch Set 3:\n\nthis patch will unblock gates for eslint-config-openstack project:\nhttps://review.opendev.org/c/openstack/eslint-config-openstack/+/752227","accounts_in_message":[],"_revision_number":3},{"id":"08f71e058d447b6933776dc40cc0f257e4d76d15","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-02-03 18:40:56.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"372ca67813a8b84e18990fd78a7c28e410682098","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-02-04 13:00:14.000000000","message":"Removed reviewer Balazs Gibizer.","accounts_in_message":[],"_revision_number":3},{"id":"d89c396d8956e9f024359312a6fd2b3deb46acc7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-10 00:38:31.000000000","message":"Patch Set 3: Code-Review+1\n\nwe should try nodejs12 also for future preparation","accounts_in_message":[],"_revision_number":3},{"id":"ce4c2be6161006b0de8bc7b3e17fbdadd71d17b0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-02-10 04:46:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cb9f0946034556404a8ce6682f586aab329929f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 04:47:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86776a215eb033bec463fe39c4694d625c26faaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 04:54:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c735a57d8ed24aee92dc7b1b5615270b : SUCCESS in 2m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/98e4cea953404187b895d9f3747b1406 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b9c452575521b39fad4072d93faa6dcdc231b1fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 04:54:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8e74f3dc56f44343bc82d0520542fc42aa11e4b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 04:58:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/49a427fb68b14a179ba2c1c62f1b1c89 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dd440c6ca47e432769b1d0b7db3aa70b4cd29c9c","revisions":{"0ea3a3f38e74bd3dabe910768cc6dc0290168b3a":{"kind":"REWORK","_number":1,"created":"2020-10-09 13:38:36.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/01/757101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/01/757101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/1"}}},"commit":{"parents":[{"commit":"24d196ec3583ca390eec9065573138f22a253827","subject":"Make tox globally available for openstack-tox-validate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/24d196ec3583ca390eec9065573138f22a253827"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-09 13:09:14.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-09 13:38:33.000000000","tz":120},"subject":"Fix nodejs*-doc project templates","message":"Fix nodejs*-doc project templates\n\nThe nodejs-npm-run-docs job is deprecated [1] therefore it\nshould be replaced with the nodejs-run-docs job as it is\nrecommended in the documentation.\n\n[1] https://zuul-ci.org/docs/zuul-jobs/deprecated-jobs.html\n\nChange-Id: I738b833109e4caeb58bb391d79d6e63284462bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0ea3a3f38e74bd3dabe910768cc6dc0290168b3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0ea3a3f38e74bd3dabe910768cc6dc0290168b3a"}]},"branch":"refs/heads/master"},"f69742568d1455ca9a9cafdab845ae80ff4abd5a":{"kind":"REWORK","_number":2,"created":"2020-10-15 07:11:36.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/01/757101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/01/757101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/2"}}},"commit":{"parents":[{"commit":"24d196ec3583ca390eec9065573138f22a253827","subject":"Make tox globally available for openstack-tox-validate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/24d196ec3583ca390eec9065573138f22a253827"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-09 13:09:14.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-15 07:11:09.000000000","tz":120},"subject":"Fix nodejs*-doc project templates","message":"Fix nodejs*-doc project templates\n\nThe nodejs-npm-run-docs job is deprecated [1] therefore it\nshould be replaced with the nodejs-run-docs job as it is\nrecommended in the documentation.\n\n[1] https://zuul-ci.org/docs/zuul-jobs/deprecated-jobs.html\n\nChange-Id: I738b833109e4caeb58bb391d79d6e63284462bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f69742568d1455ca9a9cafdab845ae80ff4abd5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f69742568d1455ca9a9cafdab845ae80ff4abd5a"}]},"branch":"refs/heads/master"},"dd440c6ca47e432769b1d0b7db3aa70b4cd29c9c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-15 07:59:55.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/01/757101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/01/757101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/757101/3"}}},"commit":{"parents":[{"commit":"24d196ec3583ca390eec9065573138f22a253827","subject":"Make tox globally available for openstack-tox-validate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/24d196ec3583ca390eec9065573138f22a253827"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-09 13:09:14.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2020-10-15 07:46:22.000000000","tz":120},"subject":"Create nodejs10-doc project templates","message":"Create nodejs10-doc project templates\n\nThe nodejs-npm-run-docs job is deprecated [1].\n\nThis patch introduces a new nodejs10-doc project template\nthat uses the replacement for nodejs-npm-run-docs\n- nodejs-run-docs.\n\n[1] https://zuul-ci.org/docs/zuul-jobs/deprecated-jobs.html\n\nChange-Id: I738b833109e4caeb58bb391d79d6e63284462bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/dd440c6ca47e432769b1d0b7db3aa70b4cd29c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/dd440c6ca47e432769b1d0b7db3aa70b4cd29c9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
