)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~760949","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I12fbe7fb894aff72b362264bdf668e9486c6c92b","project":"openstack/openstack-zuul-jobs","branch":"master","hashtags":[],"change_id":"I12fbe7fb894aff72b362264bdf668e9486c6c92b","subject":"DNM testing linter changes","status":"ABANDONED","created":"2020-11-02 15:57:05.000000000","updated":"2020-11-02 23:26:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"325f23d55804f13bffbc423ee1fe94436677852c","_number":760949,"virtual_id_number":760949,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-11-02 16:37:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-02 16:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7979611b302de02bcff5051be77f6fac65abea99","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-02 15:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d63164569100d52d14399638888886e30a682b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 16:04:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bc60f9b1f007416e94f32b7ff4bd764e : SUCCESS in 3m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/56cd5c6e75244be6910d032d6ab30e35 : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"345d89069dc685071b06cc075d35c34e7c7246cf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-02 16:10:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8e8b99af947150a6ac7aa5347ac321563fa6815c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 16:22:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bd75e4d637e54406ac344bb381e44578 : SUCCESS in 2m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4f6fb8ce28014d80a6e83e54006bf5ae : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"fce84b0ef5690f50fca677927b250c292150d85d","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-02 16:27:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60b62f98c0cccaab30df6f29d21a6e62cdefe3d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 16:37:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2af344cba3a34747be67d854d8df04da : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ed1e57f99bc4f0b8c5631818cd8043e : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"325f23d55804f13bffbc423ee1fe94436677852c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-11-02 23:26:27.000000000","message":"Abandoned\n\nThe parent change has merged and this change did what it could to show the parent worked.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76e5ce18d2c10f6b2224bfcbf23ae648c99ff0c0","revisions":{"a93261ab3cf4b5316cea4031fa3298c5ee1a497a":{"kind":"REWORK","_number":1,"created":"2020-11-02 15:57:05.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/49/760949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/760949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/1"}}},"commit":{"parents":[{"commit":"885e48c1b64e27373a880a3ad0e3a3649a388f5e","subject":"Refreshes calling of ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/885e48c1b64e27373a880a3ad0e3a3649a388f5e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-11-02 15:56:55.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-11-02 15:56:55.000000000","tz":-480},"subject":"DNM testing linter changes","message":"DNM testing linter changes\n\nChange-Id: I12fbe7fb894aff72b362264bdf668e9486c6c92b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a93261ab3cf4b5316cea4031fa3298c5ee1a497a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a93261ab3cf4b5316cea4031fa3298c5ee1a497a"}]},"branch":"refs/heads/master"},"ba1efcc8845fbffc5db00e13098733ff543df410":{"kind":"REWORK","_number":2,"created":"2020-11-02 16:10:28.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/49/760949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/760949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/2"}}},"commit":{"parents":[{"commit":"885e48c1b64e27373a880a3ad0e3a3649a388f5e","subject":"Refreshes calling of ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/885e48c1b64e27373a880a3ad0e3a3649a388f5e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-11-02 15:56:55.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-11-02 16:10:22.000000000","tz":-480},"subject":"DNM testing linter changes","message":"DNM testing linter changes\n\nChange-Id: I12fbe7fb894aff72b362264bdf668e9486c6c92b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ba1efcc8845fbffc5db00e13098733ff543df410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ba1efcc8845fbffc5db00e13098733ff543df410"}]},"branch":"refs/heads/master"},"76e5ce18d2c10f6b2224bfcbf23ae648c99ff0c0":{"kind":"REWORK","_number":3,"created":"2020-11-02 16:27:57.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/49/760949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/760949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/760949/3"}}},"commit":{"parents":[{"commit":"885e48c1b64e27373a880a3ad0e3a3649a388f5e","subject":"Refreshes calling of ansible-lint","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/885e48c1b64e27373a880a3ad0e3a3649a388f5e"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-11-02 15:56:55.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-11-02 16:27:51.000000000","tz":-480},"subject":"DNM testing linter changes","message":"DNM testing linter changes\n\nChange-Id: I12fbe7fb894aff72b362264bdf668e9486c6c92b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/76e5ce18d2c10f6b2224bfcbf23ae648c99ff0c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/76e5ce18d2c10f6b2224bfcbf23ae648c99ff0c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
