)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~800554","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Icc98e4f231413365af6a4f2270282565335e343c","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"deprecated-repo-cleanup","hashtags":[],"change_id":"Icc98e4f231413365af6a4f2270282565335e343c","subject":"Define publish-to-pypi-stable-only template","status":"MERGED","created":"2021-07-12 20:13:47.000000000","updated":"2021-07-13 23:36:27.000000000","submitted":"2021-07-13 23:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800554-deprecated-repo-cleanup","meta_rev_id":"15ffc25c1c778b47bb7c05c29ab01fb7431fc5e8","_number":800554,"virtual_id_number":800554,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-13 23:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2021-07-13 23:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-13 23:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-13 23:26:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-12 20:14:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-12 20:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-13 23:16:18.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-07-13 23:26:32.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"5900aafedc87dccb9e8857f646e993919cab7318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 20:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d7a81f8723dd750fd04fb80bab1241a1ced605a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 20:14:05.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027test-release-openstack\u0027]\n\nThe error appears in the following project-template stanza:\n\n  project-template:\n      name: publish-to-pypi-stable-only\n      description: |\n        Publish a Python package to PyPI, then send release announcement\n        emails and propose updates to upper-constraints as needed. This\n        template is supposed to be used for deliverables having stable\n        branch release only, means deprecated deliverables who still\n        support stable release.\n      check:\n        jobs:\n          - test-release-openstack:\n            # NOTE(gmann): Do not run the release job on master gate.\n            # This template is supposed to be used for stable only release.\n            branches: ^(?!master).*$\n      gate:\n        jobs:\n          - test-release-openstack:\n            branches: ^(?!master).*$\n      pre-release:\n        jobs:\n          - release-openstack-python:\n              branches: ^(?!master).*$\n          - announce-release:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n          - propose-update-constraints:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n      release:\n        jobs:\n          - release-openstack-python:\n              branches: ^(?!master).*$\n          - announce-release:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n          - propose-update-constraints:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master\", line 194, column 3","accounts_in_message":[],"_revision_number":1},{"id":"498ea41e70be8e371f7f23cb2529aa86c355e3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 20:14:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][0][\u0027test-release-openstack\u0027]\n\nThe error appears in the following project-template stanza:\n\n  project-template:\n      name: publish-to-pypi-stable-only\n      description: |\n        Publish a Python package to PyPI, then send release announcement\n        emails and propose updates to upper-constraints as needed. This\n        template is supposed to be used for deliverables having stable\n        branch release only, means deprecated deliverables who still\n        support stable release.\n      check:\n        jobs:\n          - test-release-openstack:\n            # NOTE(gmann): Do not run the release job on master gate.\n            # This template is supposed to be used for stable only release.\n            branches: ^(?!master).*$\n      gate:\n        jobs:\n          - test-release-openstack:\n            branches: ^(?!master).*$\n      pre-release:\n        jobs:\n          - release-openstack-python:\n              branches: ^(?!master).*$\n          - announce-release:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n          - propose-update-constraints:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n      release:\n        jobs:\n          - release-openstack-python:\n              branches: ^(?!master).*$\n          - announce-release:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n          - propose-update-constraints:\n              branches: ^(?!master).*$\n              dependencies:\n                - release-openstack-python\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master\", line 194, column 3","accounts_in_message":[],"_revision_number":1},{"id":"b484b6b39ae32bdb35a79b50f9140d1817dd88d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 20:15:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7199c232567aa3f8677bb5899566038575fe6418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-12 20:22:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/df557f93ce7b4d339bd29b548e2a5c9f : SUCCESS in 3m 46s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b106ddbcc15e4efe931f401c2bc8367f : SUCCESS in 3m 25s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/d7f3356a4d6a4b52a50522fd80410180 : SUCCESS in 2m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/922bb95135624a14b332e7bcb91a7d06 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"874b6a83f8455ee71d2ddc104d9e3a22f5e3dc46","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-12 20:23:49.000000000","message":"Topic set to deprecated-repo-cleanup","accounts_in_message":[],"_revision_number":2},{"id":"91055d07faeb0b95addf117d2fa29e2ba46aeef2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-13 23:03:47.000000000","message":"Patch Set 2:\n\nthis template is used in https://review.opendev.org/c/openstack/project-config/+/800558/1","accounts_in_message":[],"_revision_number":2},{"id":"0914e32a3a18883f36085fb560e07e8bd8ae57dc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2021-07-13 23:16:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6d79ed422192f773b4a3af0acdad551ee2098132","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-07-13 23:26:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0aeb3c8281afb8a37d45d5e4a7e8793bc70b8e5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 23:26:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f87945d5fa1df2c61ca3273ffbf94de92403d43d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 23:34:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2f475b9ef0564d98b01cca885fe017d8 : SUCCESS in 3m 09s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c9a480ec4024ee395fe507de97f093c : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2a51389f258dad4d53d28467b55f42488322acdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 23:34:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"15ffc25c1c778b47bb7c05c29ab01fb7431fc5e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 23:36:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/8ab60add854d480598843e7a238f4d8e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bee50a2b64cc16748a8b972848365cbbfa094c75","revisions":{"fda51e8ca6822b27bddcd6cc785eb2a83752e3fc":{"kind":"REWORK","_number":1,"created":"2021-07-12 20:13:47.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/800554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/54/800554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/1"}}},"commit":{"parents":[{"commit":"b8883fa53b0b32fde2a385bb44d24c4b21b90e2c","subject":"Merge \"Remove ocata from periodic job template of neutron and ceilometer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b8883fa53b0b32fde2a385bb44d24c4b21b90e2c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-12 20:09:57.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-12 20:09:57.000000000","tz":-300},"subject":"Define publish-to-pypi-stable-only template","message":"Define publish-to-pypi-stable-only template\n\nDeprecated deliverables only do the release on\nstable branches only not on master. But existing\npublish-to-pypi template does not have the branch\nmatcher to exclude the master gate from running the\nrelease jobs.\n\nThis commit defines a new template for such deprecated\nrepo publish-to-pypi-stable-only which exclude the master\nbranch to run the release jobs.\n\nThis new template will be used in all the deprected repo so\nthat if needed we can update the master conent which is README.rst\nfor example: https://review.opendev.org/c/openstack/neutron-fwaas/+/800140\n\nChange-Id: Icc98e4f231413365af6a4f2270282565335e343c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/fda51e8ca6822b27bddcd6cc785eb2a83752e3fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/fda51e8ca6822b27bddcd6cc785eb2a83752e3fc"}]},"branch":"refs/heads/master"},"bee50a2b64cc16748a8b972848365cbbfa094c75":{"kind":"REWORK","_number":2,"created":"2021-07-12 20:15:18.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/800554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/54/800554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/54/800554/2"}}},"commit":{"parents":[{"commit":"b8883fa53b0b32fde2a385bb44d24c4b21b90e2c","subject":"Merge \"Remove ocata from periodic job template of neutron and ceilometer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b8883fa53b0b32fde2a385bb44d24c4b21b90e2c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-12 20:09:57.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2021-07-12 20:15:11.000000000","tz":0},"subject":"Define publish-to-pypi-stable-only template","message":"Define publish-to-pypi-stable-only template\n\nDeprecated deliverables only do the release on\nstable branches only not on master. But existing\npublish-to-pypi template does not have the branch\nmatcher to exclude the master gate from running the\nrelease jobs.\n\nThis commit defines a new template for such deprecated\nrepo publish-to-pypi-stable-only which exclude the master\nbranch to run the release jobs.\n\nThis new template will be used in all the deprected repo so\nthat if needed we can update the master conent which is README.rst\nfor example: https://review.opendev.org/c/openstack/neutron-fwaas/+/800140\n\nChange-Id: Icc98e4f231413365af6a4f2270282565335e343c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bee50a2b64cc16748a8b972848365cbbfa094c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bee50a2b64cc16748a8b972848365cbbfa094c75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
