)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~801008","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I5310983bed6c5b1618b7e131996e09571be9f13d","project":"openstack/openstack-zuul-jobs","branch":"master","hashtags":[],"change_id":"I5310983bed6c5b1618b7e131996e09571be9f13d","subject":"Remove retired neutron-lbaas and django-openstack-auth from job","status":"MERGED","created":"2021-07-15 23:07:52.000000000","updated":"2021-07-27 09:08:31.000000000","submitted":"2021-07-27 09:05:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801008","meta_rev_id":"991e3a571f13b8363bd7c3da04518203381340d0","_number":801008,"virtual_id_number":801008,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-27 09:05:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2021-07-27 08:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2021-07-27 08:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-15 23:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 08:51:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-07-22 12:15:57.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-26 05:08:32.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-07-27 08:15:46.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"}],"messages":[{"id":"893c9796fbe0340faf816ad6eae7947d60e713c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-07-15 23:07:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ca19b06f440da2b9ac9b4121051df7d65fdd122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 23:18:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f4228ad23dfe4e6395e79bc38f18c7d7 : SUCCESS in 4m 38s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/fc2214a6e99b4541ac0e009faa25c584 : SUCCESS in 5m 07s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/6e667c143f2941d2913265c7ef688331 : SUCCESS in 2m 21s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/998e483d405c4854bc89d21d79aa2654 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e8a648eaa2dc42a7340251f2cd546d0584961fcc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-20 08:51:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b14be6879ad020ba378670abe176e0e015302d96","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-22 12:15:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b2f3bab2e724eb0b91c8140b31d72bdc28e989b8","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-07-26 05:08:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c59a0fe3fa57d12810c3f4c1037ba05b3929172","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-27 06:24:15.000000000","message":"Patch Set 1: Workflow+1\n\ntaking the two +1s as a +2, let\u0027s move forward...","accounts_in_message":[],"_revision_number":1},{"id":"34fdb0417920c423428ac4defa849895220d9fcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 06:24:42.000000000","message":"Patch Set 1: Verified-2\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  Unknown projects: openstack/django_openstack_auth\n\nThe error appears in the following job stanza:\n\n  job:\n      name: legacy-dsvm-base\n      abstract: true\n      description: |\n        Base job (xenial) for autoconverted legacy devstack-gate jobs\n  \n        This job runs devstack-gate with as few changes as possible and\n        may be used by jobs which have been automatically converted as\n        part of the migration to Zuul v3. This job runs on Xenial for\n        stable/ocata, pike, queens and rocky. This job is prepared to\n        make sure all stable branches before stable/stein will keep\n        running on xenial.\n      nodeset: devstack-single-node\n      branches:\n        - stable/ocata\n        - stable/pike\n        - stable/queens\n        - stable/rocky\n      pre-run: playbooks/legacy/pre.yaml\n      required-projects:\n        - openstack/devstack\n        - openstack/devstack-gate\n        - openstack/tripleo-ci\n        - openstack/ceilometer\n        - openstack/ceilometermiddleware\n        - openstack/cinder\n        - openstack/django_openstack_auth\n        - openstack/glance\n        - openstack/glance_store\n        - openstack/heat\n        - openstack/heat-cfntools\n        - openstack/heat-templates\n        - openstack/horizon\n        - openstack/keystone\n        - openstack/keystoneauth\n        - openstack/keystonemiddleware\n        - openstack/manila\n        - openstack/manila-ui\n        - openstack/neutron\n        - openstack/neutron-fwaas\n        - openstack/neutron-vpnaas\n        - openstack/nova\n        - openstack/octavia\n        - openstack/os-apply-config\n        - openstack/os-brick\n        - openstack/os-client-config\n        - openstack/os-collect-config\n        - openstack/os-net-config\n        - openstack/os-refresh-config\n        - openstack/osc-lib\n        # NOTE(mriedem): The openstack/placement repo is new in Stein and will\n        # be ignored on stable branches before Stein.\n        - openstack/placement\n        - openstack/requirements\n        - openstack/swift\n        - openstack/tempest\n        - openstack/tripleo-heat-templates\n        - openstack/tripleo-image-elements\n        - openstack/zaqar\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 1031, column 3","accounts_in_message":[],"_revision_number":1},{"id":"d191324b185382b2cf95c999fa0c7226e6b4bfac","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-27 07:16:09.000000000","message":"Patch Set 1:\n\nJust approved https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/800691, let\u0027s see whether that works - or whether we need to merge those two changes...","accounts_in_message":[],"_revision_number":1},{"id":"cd9b4bcc70bc338763da9e52447a2916cbde0a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 07:16:34.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  Unknown projects: openstack/django_openstack_auth\n\nThe error appears in the following job stanza:\n\n  job:\n      name: legacy-dsvm-base\n      abstract: true\n      description: |\n        Base job (xenial) for autoconverted legacy devstack-gate jobs\n  \n        This job runs devstack-gate with as few changes as possible and\n        may be used by jobs which have been automatically converted as\n        part of the migration to Zuul v3. This job runs on Xenial for\n        stable/ocata, pike, queens and rocky. This job is prepared to\n        make sure all stable branches before stable/stein will keep\n        running on xenial.\n      nodeset: devstack-single-node\n      branches:\n        - stable/ocata\n        - stable/pike\n        - stable/queens\n        - stable/rocky\n      pre-run: playbooks/legacy/pre.yaml\n      required-projects:\n        - openstack/devstack\n        - openstack/devstack-gate\n        - openstack/tripleo-ci\n        - openstack/ceilometer\n        - openstack/ceilometermiddleware\n        - openstack/cinder\n        - openstack/django_openstack_auth\n        - openstack/glance\n        - openstack/glance_store\n        - openstack/heat\n        - openstack/heat-cfntools\n        - openstack/heat-templates\n        - openstack/horizon\n        - openstack/keystone\n        - openstack/keystoneauth\n        - openstack/keystonemiddleware\n        - openstack/manila\n        - openstack/manila-ui\n        - openstack/neutron\n        - openstack/neutron-fwaas\n        - openstack/neutron-vpnaas\n        - openstack/nova\n        - openstack/octavia\n        - openstack/os-apply-config\n        - openstack/os-brick\n        - openstack/os-client-config\n        - openstack/os-collect-config\n        - openstack/os-net-config\n        - openstack/os-refresh-config\n        - openstack/osc-lib\n        # NOTE(mriedem): The openstack/placement repo is new in Stein and will\n        # be ignored on stable branches before Stein.\n        - openstack/placement\n        - openstack/requirements\n        - openstack/swift\n        - openstack/tempest\n        - openstack/tripleo-heat-templates\n        - openstack/tripleo-image-elements\n        - openstack/zaqar\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 1031, column 3","accounts_in_message":[],"_revision_number":1},{"id":"a0e4ba3bdd651b1a27a1de917fffad3e3e9ca08b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-27 07:26:51.000000000","message":"Patch Set 1:\n\nWe need to combine this and the django_openstack_auth change to get those in - anybody that can do this, please?","accounts_in_message":[],"_revision_number":1},{"id":"7e256dc766fd621131605b2c385eecc1f2ea9d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 07:27:14.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  Unknown projects: openstack/django_openstack_auth\n\nThe error appears in the following job stanza:\n\n  job:\n      name: legacy-dsvm-base\n      abstract: true\n      description: |\n        Base job (xenial) for autoconverted legacy devstack-gate jobs\n  \n        This job runs devstack-gate with as few changes as possible and\n        may be used by jobs which have been automatically converted as\n        part of the migration to Zuul v3. This job runs on Xenial for\n        stable/ocata, pike, queens and rocky. This job is prepared to\n        make sure all stable branches before stable/stein will keep\n        running on xenial.\n      nodeset: devstack-single-node\n      branches:\n        - stable/ocata\n        - stable/pike\n        - stable/queens\n        - stable/rocky\n      pre-run: playbooks/legacy/pre.yaml\n      required-projects:\n        - openstack/devstack\n        - openstack/devstack-gate\n        - openstack/tripleo-ci\n        - openstack/ceilometer\n        - openstack/ceilometermiddleware\n        - openstack/cinder\n        - openstack/django_openstack_auth\n        - openstack/glance\n        - openstack/glance_store\n        - openstack/heat\n        - openstack/heat-cfntools\n        - openstack/heat-templates\n        - openstack/horizon\n        - openstack/keystone\n        - openstack/keystoneauth\n        - openstack/keystonemiddleware\n        - openstack/manila\n        - openstack/manila-ui\n        - openstack/neutron\n        - openstack/neutron-fwaas\n        - openstack/neutron-vpnaas\n        - openstack/nova\n        - openstack/octavia\n        - openstack/os-apply-config\n        - openstack/os-brick\n        - openstack/os-client-config\n        - openstack/os-collect-config\n        - openstack/os-net-config\n        - openstack/os-refresh-config\n        - openstack/osc-lib\n        # NOTE(mriedem): The openstack/placement repo is new in Stein and will\n        # be ignored on stable branches before Stein.\n        - openstack/placement\n        - openstack/requirements\n        - openstack/swift\n        - openstack/tempest\n        - openstack/tripleo-heat-templates\n        - openstack/tripleo-image-elements\n        - openstack/zaqar\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 1031, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e6e6b10e5b5eb26580b745f439242d698996d853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-27 08:15:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f112c7e73a73228586e688b8f2efdf7bf796c5bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-27 08:16:50.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"082399f43a243ee6e25305d920724d4685dd90fa","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2021-07-27 08:19:21.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 1:\n\u003e \n\u003e We need to combine this and the django_openstack_auth change to get those in - anybody that can do this, please?\n\nDone.","accounts_in_message":[],"_revision_number":3},{"id":"61013877f8082c80a1ab791beb9212da50b3ebc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:29:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/50e2b44a23e44b98b49c0232c9c1a04f : SUCCESS in 4m 08s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b8a6d971c4074a0497782b5fe1e436cb : SUCCESS in 3m 40s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/5517cd0e57ce472e83383d99c7e8577e : SUCCESS in 3m 32s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a5679ac66d744dbab27769ff6ed16d4 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"422bde138182896a4cf8f057ade83d7d70e803cb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-07-27 08:55:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\ncleanup patch, single core approving.\n\nThanks, Vishal!","accounts_in_message":[],"_revision_number":3},{"id":"b00c0ee5b9ea9fb5b03e3431f7bad3b0dad999ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:55:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b7e2c7fccac152d38c1b490040261f9c53abdf02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:05:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/269dce88579e4848872fdfda21f25eda : SUCCESS in 3m 23s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1e236d0f72354b979ab1e49fc0cad18d : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":3},{"id":"61b93ee9b6deaf7d97f98e21789f190c696f76ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:05:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"991e3a571f13b8363bd7c3da04518203381340d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:08:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/1072f3eaae3344b39b1abec4204ace06 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a3eb3afeb3eb9eb15c1e595e02e7d88c900607f","revisions":{"46c796fc07fb7ff54ddea77bad6f746c69e8ef35":{"kind":"REWORK","_number":1,"created":"2021-07-15 23:07:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/801008/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/08/801008/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/1"}}},"commit":{"parents":[{"commit":"3c6f0ebd295c84aa1f5e8e3aec691d787c1a9884","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3c6f0ebd295c84aa1f5e8e3aec691d787c1a9884"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-15 23:07:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-15 23:07:16.000000000","tz":-300},"subject":"Remove retired neutron-lbaas from job","message":"Remove retired neutron-lbaas from job\n\nneutron-lbaas is retired now and we can\nremove it from zuul jobs.\n\nChange-Id: I5310983bed6c5b1618b7e131996e09571be9f13d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/46c796fc07fb7ff54ddea77bad6f746c69e8ef35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/46c796fc07fb7ff54ddea77bad6f746c69e8ef35"}]},"branch":"refs/heads/master"},"90465b16d07eca500d4ea669c373eac4a4bdf9f6":{"kind":"REWORK","_number":2,"created":"2021-07-27 08:15:46.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/08/801008/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/08/801008/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/2"}}},"commit":{"parents":[{"commit":"3c6f0ebd295c84aa1f5e8e3aec691d787c1a9884","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3c6f0ebd295c84aa1f5e8e3aec691d787c1a9884"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-15 23:07:16.000000000","tz":-300},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-27 08:12:38.000000000","tz":330},"subject":"Remove retired neutron-lbaas django-openstack-auth from job","message":"Remove retired neutron-lbaas django-openstack-auth from job\n\nneutron-lbaas is retired now and we can\nremove it from zuul jobs. It also retire django_openstack_auth\nin this patch to unblock the gate.\n\n`django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: I5310983bed6c5b1618b7e131996e09571be9f13d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/90465b16d07eca500d4ea669c373eac4a4bdf9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/90465b16d07eca500d4ea669c373eac4a4bdf9f6"}]},"branch":"refs/heads/master"},"7a3eb3afeb3eb9eb15c1e595e02e7d88c900607f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-27 08:16:50.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/08/801008/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/08/801008/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/08/801008/3"}}},"commit":{"parents":[{"commit":"3c6f0ebd295c84aa1f5e8e3aec691d787c1a9884","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3c6f0ebd295c84aa1f5e8e3aec691d787c1a9884"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-07-15 23:07:16.000000000","tz":-300},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2021-07-27 08:16:28.000000000","tz":330},"subject":"Remove retired neutron-lbaas and django-openstack-auth from job","message":"Remove retired neutron-lbaas and django-openstack-auth from job\n\nneutron-lbaas is retired now and we can\nremove it from zuul jobs. It also retire django_openstack_auth\nin this patch to unblock the gate.\n\n`django_openstack_auth`` code was merged into the main horizon\nrepository during the queens release. This repo is already\ndeprecated and now we can retire it. This patch removes all\nreferences of ``django_openstack_auth`` as mentioned in [1].\n\n[1] https://docs.openstack.org/project-team-guide/repository.html#retiring-a-repository\n\nChange-Id: I5310983bed6c5b1618b7e131996e09571be9f13d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/7a3eb3afeb3eb9eb15c1e595e02e7d88c900607f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/7a3eb3afeb3eb9eb15c1e595e02e7d88c900607f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
