)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~840147","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I5a4491f4ffc65e37f353ca66139d015eb097e5a6","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"openafs-centos9","hashtags":[],"change_id":"I5a4491f4ffc65e37f353ca66139d015eb097e5a6","subject":"openafs: fix 9-stream dir typo","status":"MERGED","created":"2022-05-02 02:37:40.000000000","updated":"2022-05-02 04:08:00.000000000","submitted":"2022-05-02 04:06:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840147-openafs-centos9","meta_rev_id":"1b273b06b37e1cd8a5b8e2138104cd6904be9f65","_number":840147,"virtual_id_number":840147,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 04:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-02 03:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-02 03:47:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-02 03:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e051b2b5a94992dd3e0995fcb9c33e1631925e99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-02 02:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98df474cea06a7cb1be9bf7b1e574bbb54f3a7c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 03:41:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/325a324191fe41d2aebfc86d2da1e548 : SUCCESS in 2m 54s\n- openafs-rpm-package-build-centos-7-x86 https://zuul.opendev.org/t/openstack/build/df503d98909e410c87353eef59f6dd12 : SUCCESS in 8m 49s\n- openafs-rpm-package-build-centos-8-stream-x86 https://zuul.opendev.org/t/openstack/build/4c2904a0288344e4bd555ae9c00b188a : SUCCESS in 13m 36s\n- openafs-rpm-package-build-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/f93d9bb731394c4bab0fe674d624b8f6 : SUCCESS in 12m 47s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/b307a4860f394dedb0e6fd74fa8b902c : SUCCESS in 11m 36s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f662080c4bdc4411bc98664d76e6761e : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a3c2511e61491331f32d44805c11559c29d8ef07","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-02 03:47:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea5d1a49026b9ce19878b57ddfa29d679b2b6aa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 03:47:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e939f04ea16ade8721c27424aa98d1b82a9584e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 04:06:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/07a8418e55694c12985217945625efb4 : SUCCESS in 2m 16s\n- openafs-rpm-package-build-centos-7-x86 https://zuul.opendev.org/t/openstack/build/371b478951cd4c4593af99417c696d5a : SUCCESS in 8m 42s\n- openafs-rpm-package-build-centos-8-stream-x86 https://zuul.opendev.org/t/openstack/build/539621838e2d4999bf4fef83a82bdc75 : SUCCESS in 13m 02s\n- openafs-rpm-package-build-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/8b365941ab21416b9d99ec0ce89524a5 : SUCCESS in 11m 43s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/de6d586b7f79432596ef1f532b7bbad5 : SUCCESS in 13m 38s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b52a2a3351ae4b399d0f1013716343c6 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ac1cfe6f651ece397b2f3b4e7c464819975498d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 04:06:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1b273b06b37e1cd8a5b8e2138104cd6904be9f65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 04:08:00.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/58266d68776d4d9186b38598e0c02f90 : SUCCESS in 1m 06s\n- openafs-rpm-package-build-promote https://zuul.opendev.org/t/openstack/build/17c2569ac6a440f28903feed177d9522 : FAILURE in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34300b6526eb8ab13c30e0dc8d5bf746551ad7f2","revisions":{"34300b6526eb8ab13c30e0dc8d5bf746551ad7f2":{"kind":"REWORK","_number":1,"created":"2022-05-02 02:37:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/47/840147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/47/840147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/47/840147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/47/840147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/47/840147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/47/840147/1"}}},"commit":{"parents":[{"commit":"1d7825865e8e0785c70835026f980c1ce99e6b8c","subject":"openafs: fix extraction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1d7825865e8e0785c70835026f980c1ce99e6b8c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-02 02:37:23.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-02 02:37:23.000000000","tz":600},"subject":"openafs: fix 9-stream dir typo","message":"openafs: fix 9-stream dir typo\n\nChange-Id: I5a4491f4ffc65e37f353ca66139d015eb097e5a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/34300b6526eb8ab13c30e0dc8d5bf746551ad7f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/34300b6526eb8ab13c30e0dc8d5bf746551ad7f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
