)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~847862","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Icd1b531de30a34780a340ea02209f49eb925de36","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"grafana-buildset-soft","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-06-27 23:07:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd1b531de30a34780a340ea02209f49eb925de36","subject":"project-config-grafana: make opendev-buildset-registry soft dep","status":"MERGED","created":"2022-06-27 22:42:09.000000000","updated":"2022-06-27 23:09:42.000000000","submitted":"2022-06-27 23:07:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847862-grafana-buildset-soft","meta_rev_id":"55bbb88b294b5cdd3f8376ede739057c06c625d4","_number":847862,"virtual_id_number":847862,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-27 23:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-06-27 22:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-27 22:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-06-27 22:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-27 22:43:05.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-06-27 22:53:52.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-06-27 22:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9efb29f59773f63db23c6f5426b9b6dfcaa720fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-27 22:42:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"707faa19222ebc92e26a441f2c202ada5a93b802","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-06-27 22:43:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd13e62c40bbcef797c354494ba8b44c2d07322c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-27 22:53:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12fd8cd368e0d30ed657363876a1e27bd55563e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 22:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4afba660ce0a4ea19425681914946ac7 : SUCCESS in 3m 21s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c605c920fb9549de8dbd6cbe0068147a : SUCCESS in 16m 14s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/db9e48ea132043b0b7743739d5862061 : SUCCESS in 11m 53s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/74d59ed14aeb4b2ebdfc2d02e3d3e192 : SUCCESS in 2m 41s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/6de5bfb195224695b130b72934b29883 : SUCCESS in 2m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a431d22e7709455f8fb0c6ee4e8e5766 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4a5a02be2f1e67d7e9257a6e3b89de24688bac31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:00:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c74dc7bf76541a9b2658937a9e68929f3bf48231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:07:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5e1547c7d745467c95cb2faf92393e72 : SUCCESS in 3m 30s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa7b5d039eef4fafbaae6464c08ecf04 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"41a276a0a8506deffce207dd5b0ae56aa9eafa94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:07:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55bbb88b294b5cdd3f8376ede739057c06c625d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:09:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/9469398026414b36856758b002e21c4e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ace49f32d923430c8dab0597f7c0af0eb3da9a96","revisions":{"ace49f32d923430c8dab0597f7c0af0eb3da9a96":{"kind":"REWORK","_number":1,"created":"2022-06-27 22:42:09.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/847862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/62/847862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/62/847862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/62/847862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/62/847862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/62/847862/1"}}},"commit":{"parents":[{"commit":"fb7158602f2c01d2e44ceb65e34f2aa3e70203f3","subject":"Merge \"grafana : import and screenshot project-config graphs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/fb7158602f2c01d2e44ceb65e34f2aa3e70203f3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-27 22:40:53.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-27 22:40:53.000000000","tz":600},"subject":"project-config-grafana: make opendev-buildset-registry soft dep","message":"project-config-grafana: make opendev-buildset-registry soft dep\n\nThis job was meant to work by either pulling through the registry for\nspeculative testing, or for project-config just running on released\nimages.  Make this soft so project-config doens\u0027t have to start a\nregistry.\n\nChange-Id: Icd1b531de30a34780a340ea02209f49eb925de36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ace49f32d923430c8dab0597f7c0af0eb3da9a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ace49f32d923430c8dab0597f7c0af0eb3da9a96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
