)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~848039","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ib84dfc65f76c021955f109912774408fa2393d07","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"grafana-fail","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-06-30 01:36:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib84dfc65f76c021955f109912774408fa2393d07","subject":"[dnm] grafana fail","status":"ABANDONED","created":"2022-06-29 00:06:42.000000000","updated":"2022-06-30 01:36:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3bb32d44ba04513e415e9c077a442bc328a02981","_number":848039,"virtual_id_number":848039,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-29 00:43:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 00:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"370c0ebe88db25cb2b7224d01e888423da932b1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-29 00:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17f03c56d558676611c5dd6df9225803a25b25f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-29 00:08:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9e5446a1be8d96c7eb2604a58ee0f23a855e1c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-29 00:15:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a00605263b8a0f205e25c081dcb85e83b7d3b63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 00:43:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e1d58241bc6d4077ab4df52dab4baf6d : SUCCESS in 2m 58s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/51791b9438ad450db79cf4368a761a4e : SUCCESS in 21m 38s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/4053619aa10e4bac967ed44436c697fc : FAILURE in 11m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/95dc64c6df714c7ea48ff34b50518177 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":3},{"id":"d61437d486cb883d9fbada54303c7fb5e596015a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-29 00:48:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1232a710e9016eaeb03123aad043a70fa4f6e4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 01:08:43.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1ffe26d3a0c74c8a86423024ec23b5ef : SUCCESS in 4m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c008340547774380861e540581b3f5a6 : SUCCESS in 15m 03s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/b0583406b86a46c5a298f21aaf11c529 : FAILURE in 11m 34s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9c4f66f4987a449f8183b2eb217c5075 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3bb32d44ba04513e415e9c077a442bc328a02981","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-30 01:36:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"196df0e7bc0b3512448f36f4441b549151f87fcf","revisions":{"ec5dd0cf06071ca8c0ad86fe814facc98ce5d1ad":{"kind":"REWORK","_number":1,"created":"2022-06-29 00:06:42.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/39/848039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/39/848039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/1"}}},"commit":{"parents":[{"commit":"45d96f730ebb7b96603b70340ead81f04119d80d","subject":"Revert \"project-config-grafana: make opendev-buildset-registry soft dep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/45d96f730ebb7b96603b70340ead81f04119d80d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-29 00:06:17.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-29 00:06:17.000000000","tz":600},"subject":"[dnm] grafana fail","message":"[dnm] grafana fail\n\nChange-Id: Ib84dfc65f76c021955f109912774408fa2393d07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ec5dd0cf06071ca8c0ad86fe814facc98ce5d1ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ec5dd0cf06071ca8c0ad86fe814facc98ce5d1ad"}]},"branch":"refs/heads/master"},"72fe2564343088a2082769de1335ee272893f0df":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-29 00:08:45.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/39/848039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/39/848039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/2"}}},"commit":{"parents":[{"commit":"45d96f730ebb7b96603b70340ead81f04119d80d","subject":"Revert \"project-config-grafana: make opendev-buildset-registry soft dep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/45d96f730ebb7b96603b70340ead81f04119d80d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-29 00:06:17.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-29 00:08:29.000000000","tz":600},"subject":"[dnm] grafana fail","message":"[dnm] grafana fail\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/833213\nChange-Id: Ib84dfc65f76c021955f109912774408fa2393d07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/72fe2564343088a2082769de1335ee272893f0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/72fe2564343088a2082769de1335ee272893f0df"}]},"branch":"refs/heads/master"},"196df0e7bc0b3512448f36f4441b549151f87fcf":{"kind":"REWORK","_number":3,"created":"2022-06-29 00:15:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/39/848039/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/39/848039/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/39/848039/3"}}},"commit":{"parents":[{"commit":"45d96f730ebb7b96603b70340ead81f04119d80d","subject":"Revert \"project-config-grafana: make opendev-buildset-registry soft dep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/45d96f730ebb7b96603b70340ead81f04119d80d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-29 00:06:17.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-29 00:14:55.000000000","tz":600},"subject":"[dnm] grafana fail","message":"[dnm] grafana fail\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/833213\nChange-Id: Ib84dfc65f76c021955f109912774408fa2393d07\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/196df0e7bc0b3512448f36f4441b549151f87fcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/196df0e7bc0b3512448f36f4441b549151f87fcf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
