)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~850752","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I90023d9cdf763bc86235e050a917203945bc39bc","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"eol_vpnaas_t_u","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-02-07 08:32:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90023d9cdf763bc86235e050a917203945bc39bc","subject":"VPNAAS, FWAAS: remove old jobs from legacy jobs","status":"MERGED","created":"2022-07-22 13:42:36.000000000","updated":"2023-02-07 08:34:44.000000000","submitted":"2023-02-07 08:32:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850752-eol_vpnaas_t_u","meta_rev_id":"6a955f0053567b067ab930c504c906c773db62c0","_number":850752,"virtual_id_number":850752,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-07 08:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-11-29 07:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2023-02-07 08:22:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 13:48:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 07:34:58.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"b743821fa068c12eb8552ae5a610a0a2e90f0d39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-22 13:42:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a5c71194a1f1f3af7bc976f06cfba3beef104bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 13:48:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fab406e1970140608ddfb4dacc91b001\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f10eb4e2fe6f4a52a14f471ace1ae3b4 : SUCCESS in 3m 18s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1b8b8e5f6f9542d08d6d7c59f9e2e5ac : SUCCESS in 4m 23s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b673c861054f4f88bb6c64bb552d8f3b : SUCCESS in 3m 27s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/a543ed98a93f450da116702889941400 : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7bf768b2c99440248151e80f1a4b6409 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e6171d217f3a83001f9fccec887ec71e0bf7564d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-29 07:34:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08e97356d576ce9c13a1b1b8acb3775df5d40d1e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-02-07 08:22:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d69aed18d9a2c4cbd6dd0a28eb108e0c0358f45a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:22:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4f856c388f271045cd63681cb51a30fc8a517cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:32:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/188f7f9f08014754918bf007ab5efdcb\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d5aa8651c8bc4310b38fabd0cd094e37 : SUCCESS in 2m 09s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cfcae584eae24a4a87992f64973aaa57 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9ee55d972bdbce55ce591b2b03d737f7597f7a53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:32:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a955f0053567b067ab930c504c906c773db62c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 08:34:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0acf5203285f411c812f43d1c9216714\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/c01d0d5ff14f4c9db29235a7c0016772 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d5fa22af554fa3644bf4db8d385ae57dbf0367d","revisions":{"9d5fa22af554fa3644bf4db8d385ae57dbf0367d":{"kind":"REWORK","_number":1,"created":"2022-07-22 13:42:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/52/850752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/52/850752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/52/850752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/52/850752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/52/850752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/52/850752/1"}}},"commit":{"parents":[{"commit":"96991ebe40589d08b0fd1a3d4cab4239320a1646","subject":"project-config-grafana: collect container logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/96991ebe40589d08b0fd1a3d4cab4239320a1646"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-22 13:40:30.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-22 13:40:30.000000000","tz":120},"subject":"VPNAAS, FWAAS: remove old jobs from legacy jobs","message":"VPNAAS, FWAAS: remove old jobs from legacy jobs\n\nneutron-vpnaas Train and Ussuri branches are no more maintained (see\n[1]), so the EOL process has been started on them (see [2]), and as a\nconsequence for these and for older branches vpnaas can be removed from\nlegacy job requirements.\nneutron-fwaas EOL-d these branches already so it can be also removed.\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2022-May/028728.html\n[2]: https://review.opendev.org/c/openstack/releases/+/843167\n\nChange-Id: I90023d9cdf763bc86235e050a917203945bc39bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/9d5fa22af554fa3644bf4db8d385ae57dbf0367d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/9d5fa22af554fa3644bf4db8d385ae57dbf0367d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[]}
