)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~859448","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I9eac4d68ed15e0dcb4685d13e5738694f1325e71","project":"openstack/openstack-zuul-jobs","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-28 13:10:40.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-09-28 13:10:40.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-09-28 13:10:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9eac4d68ed15e0dcb4685d13e5738694f1325e71","subject":"Add stable/zed to periodic-stable templates","status":"MERGED","created":"2022-09-27 15:54:18.000000000","updated":"2022-09-28 13:12:39.000000000","submitted":"2022-09-28 13:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859448","meta_rev_id":"a2836e30c0f45ea3b9aa9fc18f0700e3fda4fee9","_number":859448,"virtual_id_number":859448,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-28 13:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-28 11:51:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2022-09-28 12:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2022-09-28 12:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-27 15:57:19.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-09-27 16:05:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-09-27 16:05:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-09-27 16:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-28 11:51:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c11468e731cdf40c8bd79584e613032bb96bedbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-27 15:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d59a3c198756b1a57d935499821f70b165dad22","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-27 15:57:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"df3de7f3f8e85238b2bc6450f5a6df63744ee9ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 16:06:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2b771ae20647bb866bd5167897f0a1\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/17187d78e62548d6b55d606ce6d850fa : SUCCESS in 3m 06s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/619a1ea0faea4b29a6d7b1914a3c6e8d : SUCCESS in 3m 56s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b274dcc47b704cb4a6a8da5af589be50 : SUCCESS in 2m 51s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/bf6ab3edb11c45df8e07ceffc4c50d62 : SUCCESS in 2m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1867b6c8dc244629b9949645ef14ce6a : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6f930687805d414d6fc99a3cf57b8387284372f4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-09-28 02:03:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7d82ef0b9a8662eff90c36efce29a3bec028bee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-28 07:03:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"239b8c6c2ef37b1c6e196dd79dc8f5c6387085a0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-28 07:04:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c2c9a226635ee109e0e7e6d4c26c2e957ca865d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 07:10:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/758357a6e11f4b6399a6e435d42ac0ea\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bde554ddcbd54ad294c938a42d7dbc36 : SUCCESS in 3m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/276ad7dfaa2440d0baec2bf57bfc961f : SUCCESS in 4m 06s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b30b9a64fdc94567ab1c15b3d216a28e : SUCCESS in 3m 22s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/cc0dac4ad7404f199a22925b3ca04394 : SUCCESS in 2m 49s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d02faa6cbcfe4f86bb88e3ea97c852f5 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"60796d3bc5478d342a4706559fce2f993f870150","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-28 11:51:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f10c6793b5700d1f4191f1b222d82f96efa92656","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-28 12:59:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56ff53a715480f61e6786d7766e49c7ff5c9085e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 12:59:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a11c5357c88aaa4fff139ac80d8b7880ddc21e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:10:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/299d9c582d1246f5bdc20506a3633475\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f036389bef1b4c96bc4412e7d7c3a411 : SUCCESS in 5m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/91c1ee7392204cac9a59e74d28312849 : SUCCESS in 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"dbcbfda4ba1534e1e3e5956915087788a5ea82d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:10:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a2836e30c0f45ea3b9aa9fc18f0700e3fda4fee9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 13:12:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbf9f32625b478d9d31713a43bf1670\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/c2abfbfd436e4a02829b42e32376d7d7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e63f2b904d291c2cd4fe716121abf71e3ad1abc","revisions":{"316ec52ea052fcd485327ccbd71a735de26bf77a":{"kind":"REWORK","_number":1,"created":"2022-09-27 15:54:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/859448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/48/859448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/1"}}},"commit":{"parents":[{"commit":"52d4006a2c53cf805dabf07f4e743d01316e0380","subject":"Merge \"Add missing openstack-tox-py310-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/52d4006a2c53cf805dabf07f4e743d01316e0380"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-27 15:53:12.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-27 15:53:12.000000000","tz":120},"subject":"Add stable/zed to periodic-stable templates","message":"Add stable/zed to periodic-stable templates\n\nAdd the stable/zed branch to the list of branches for the\nperiodic stable job templates.\n\nChange-Id: I9eac4d68ed15e0dcb4685d13e5738694f1325e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/316ec52ea052fcd485327ccbd71a735de26bf77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/316ec52ea052fcd485327ccbd71a735de26bf77a"}]},"branch":"refs/heads/master"},"1e63f2b904d291c2cd4fe716121abf71e3ad1abc":{"kind":"REWORK","_number":2,"created":"2022-09-28 07:03:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/48/859448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/48/859448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/48/859448/2"}}},"commit":{"parents":[{"commit":"52d4006a2c53cf805dabf07f4e743d01316e0380","subject":"Merge \"Add missing openstack-tox-py310-arm64 job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/52d4006a2c53cf805dabf07f4e743d01316e0380"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-27 15:53:12.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-28 07:02:46.000000000","tz":120},"subject":"Add stable/zed to periodic-stable templates","message":"Add stable/zed to periodic-stable templates\n\nAdd the stable/zed branch to the list of branches for the\nperiodic stable job templates.\n\nChange-Id: I9eac4d68ed15e0dcb4685d13e5738694f1325e71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1e63f2b904d291c2cd4fe716121abf71e3ad1abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1e63f2b904d291c2cd4fe716121abf71e3ad1abc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
