)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~863012","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ib1a55d624c72aa773ab03b736efed0699f35e946","project":"openstack/openstack-zuul-jobs","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-10-31 12:07:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1a55d624c72aa773ab03b736efed0699f35e946","subject":"Run build-openstack-puppet-tarball on Focal","status":"MERGED","created":"2022-10-31 06:13:24.000000000","updated":"2022-10-31 12:10:22.000000000","submitted":"2022-10-31 12:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863012","meta_rev_id":"730a6f8db5b2027d564f10e39cc6774cac728fd5","_number":863012,"virtual_id_number":863012,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-31 12:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2022-10-31 07:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-31 12:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-31 12:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-31 06:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 07:02:09.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2022-10-31 12:00:58.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0ea9e609361d916ace37cb54260cee8c8364f25c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 06:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01831cdc50469580de000034f29d7194981ce3b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 06:18:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2fdedab5f2d4cc2bb2077fae29ed369\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/10ce8a834bdd4ba7a0cf0830ed0de7a1 : SUCCESS in 2m 27s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/888a343e8de44a34a7409700ffdd9a69 : SUCCESS in 2m 26s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/9c98844ec4884cf8b4e4088b721bee64 : SUCCESS in 3m 03s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/9347d1ddfd934bb39732aee641a7e069 : SUCCESS in 2m 23s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/91747fff8fc8429480af42efc050d4ce : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"257448f82a3410b340a8cbcbd8a82bebb242374a","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 07:02:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4c088f989681e1cbb164e2a56c499ee6d5b292c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-31 12:00:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis will need to be updated in the coming months, in accordance with the https://governance.openstack.org/tc/goals/selected/migrate-ci-jobs-to-ubuntu-jammy.html goal.\n\nAlso, I wouldn\u0027t be surprised if some of other jobs failed similarly since it looks like this is a result of newer Ruby on Jammy.","accounts_in_message":[],"_revision_number":1},{"id":"68c706ebaefe023ef0c72308114aa29984fa3503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:01:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e4fd5678e5c9a5ce329a019512d31808eda4909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:07:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3deb50ccda134003a1ec5eac79bbd216\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/630bf347d20e4283b2460bbf730b0ee3 : SUCCESS in 3m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/73a55b8c51fc45baab5e52c9595c8d78 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"97a166edf2a7ead0f2bc610fe0b7d74b9e9afa21","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:07:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"730a6f8db5b2027d564f10e39cc6774cac728fd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20df520f9de54d8cad8fee7837e1c5c7\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/795e0e24d8d143f6a120e144e8069798 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b20eb582a00c0286001c439efba7bce839783fdd","revisions":{"b20eb582a00c0286001c439efba7bce839783fdd":{"kind":"REWORK","_number":1,"created":"2022-10-31 06:13:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/863012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/12/863012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/863012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/863012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/863012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/12/863012/1"}}},"commit":{"parents":[{"commit":"7e7045ab92b0b9db28f24fe9a38f914f74174938","subject":"Start translations for Zed stable branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/7e7045ab92b0b9db28f24fe9a38f914f74174938"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-31 06:12:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-10-31 06:13:22.000000000","tz":540},"subject":"Run build-openstack-puppet-tarball on Focal","message":"Run build-openstack-puppet-tarball on Focal\n\nThe build-openstack-puppet-tarball job consistently fails since\nthe default node set was switched from Focal to Jammy. This change\npins the node set back to Focal until we fix the current issue.\n\nChange-Id: Ib1a55d624c72aa773ab03b736efed0699f35e946\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b20eb582a00c0286001c439efba7bce839783fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b20eb582a00c0286001c439efba7bce839783fdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
