)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~882575","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I4b2dfc8436d4238470af31f5a8d7a881fe254558","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"opendev-quay","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-05-10 20:28:06.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}},"hashtags":[],"change_id":"I4b2dfc8436d4238470af31f5a8d7a881fe254558","subject":"Pull grafyaml from quay.io","status":"MERGED","created":"2023-05-08 15:54:17.000000000","updated":"2023-05-10 20:46:45.000000000","submitted":"2023-05-10 20:45:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882575-opendev-quay","meta_rev_id":"47e34522dba91138fbdf9c314b5b995094d370ff","_number":882575,"virtual_id_number":882575,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-10 20:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-08 22:12:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-05-10 20:28:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-08 16:09:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-08 22:12:46.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"f56b63a332bb40346e19ed58b5fce0cf5e0072b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-08 15:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"592573522b6940b348cf7d09e1374ebce032baee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 16:09:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db12524638724f818f0e34d5417d3630\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/831eb04ae96b44cabec9c46634b4bd5c : SUCCESS in 2m 24s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8013d8b7a04e4174b9e34e2419455944 : SUCCESS in 14m 12s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/0ffa72efe0b948f399170ebae65af7c7 : SUCCESS in 10m 59s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a402757658a947149653b2862f979848 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1f38a1cdafb2be77de15a218f0fcd77c05b66804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-08 16:21:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bea249a6bb3a0d7790c76ddbc6b188aff772aee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 16:45:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afbb7d21764443ddb870549f520e0cb1\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d9a6a6dd75534542ad99a66c47e08036 : SUCCESS in 2m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/656aab1e992f465f864d9e934e0fc49f : SUCCESS in 19m 09s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/f74c537ebc914e568470efb0b999120b : SUCCESS in 12m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ce66442509dc416ba44e5de36cb46286 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"abcd37a54e6e25ef56a573aa9062009c19ae3cf9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-08 20:01:52.000000000","message":"Topic changed from opendev-quayio to opendev-quay","accounts_in_message":[],"_revision_number":2},{"id":"530eb3295bd98dcf17d8520b0e172c2f7bb7ac61","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-05-08 22:12:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4302452baf11882782b6d40f0022966eb43be39f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-05-10 20:28:06.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8fb86cbc8aa30f398152a3ef119ecffb39751257","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:28:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd4e254939cc8052a9990f0021488eade3c6a3a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:45:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6528e5996e3c45f3be622ab8447bd259\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/8b58cfd574ee4766bc60cd254888d5eb : SUCCESS in 2m 19s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/78d2a23610e042abbf61f2baf2203733 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"8a7e93e4b2eb1dbfa4edff9039c678574c55c8f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:45:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"47e34522dba91138fbdf9c314b5b995094d370ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 20:46:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9e75c443a05489789f694ba5bd585b7\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/156f3d59cbc74ad4801694286e857e3a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6f701cfeb53cd8f489a6b8882eafdeb7a6c1dac","revisions":{"2eb6634c82a53fbec57268ee546e0e0fc2b7f5e0":{"kind":"REWORK","_number":1,"created":"2023-05-08 15:54:17.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/75/882575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/75/882575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/1"}}},"commit":{"parents":[{"commit":"0e9aa6c80879f78b3c346631210c2ccdcf4c4d21","subject":"Merge \"Fix installation of py27 on CentOS 7/8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0e9aa6c80879f78b3c346631210c2ccdcf4c4d21"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-05-08 15:53:38.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-05-08 15:53:38.000000000","tz":-420},"subject":"Pull grafyaml from quay.io","message":"Pull grafyaml from quay.io\n\nThe grafyaml image has moved to quay.io. Ensure we are pulling it from\nthat new maintained location.\n\nDepends-On: https://review.opendev.org/c/opendev/grafyaml/+/882493\nChange-Id: I4b2dfc8436d4238470af31f5a8d7a881fe254558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2eb6634c82a53fbec57268ee546e0e0fc2b7f5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2eb6634c82a53fbec57268ee546e0e0fc2b7f5e0"}]},"branch":"refs/heads/master"},"d6f701cfeb53cd8f489a6b8882eafdeb7a6c1dac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-08 16:21:14.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/75/882575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/75/882575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/75/882575/2"}}},"commit":{"parents":[{"commit":"d4433dfe34ef0f743af812da20caa9dcb18f58ff","subject":"Exclude more broken ansile-compat library versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d4433dfe34ef0f743af812da20caa9dcb18f58ff"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-05-08 15:53:38.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-05-08 16:21:06.000000000","tz":-420},"subject":"Pull grafyaml from quay.io","message":"Pull grafyaml from quay.io\n\nThe grafyaml image has moved to quay.io. Ensure we are pulling it from\nthat new maintained location.\n\nDepends-On: https://review.opendev.org/c/opendev/grafyaml/+/882493\nChange-Id: I4b2dfc8436d4238470af31f5a8d7a881fe254558\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d6f701cfeb53cd8f489a6b8882eafdeb7a6c1dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d6f701cfeb53cd8f489a6b8882eafdeb7a6c1dac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
