)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~890432","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Icd7f2e10522c2cf69107db74823bf8b767a9e53d","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"snaps-increase-timeout","attention_set":{},"removed_from_attention_set":{"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2023-08-03 17:58:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd7f2e10522c2cf69107db74823bf8b767a9e53d","subject":"openstack-tox-snap-with-sudo: increase timeout","status":"MERGED","created":"2023-08-03 15:25:47.000000000","updated":"2023-08-03 18:00:48.000000000","submitted":"2023-08-03 17:58:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890432-snaps-increase-timeout","meta_rev_id":"2ac752edf596455cf4698417da8643b49417649a","_number":890432,"virtual_id_number":890432,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-03 17:58:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-03 17:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-03 17:49:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 15:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-03 17:49:34.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"0c06712786e7e433162023f6a9eea8e484115e30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2023-08-03 15:25:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2457301ee80fc1cdd3569058bc1d3091f376520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 15:39:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87bb4c542b3240268ae697b35b81ae1d\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e8030d8aa4de4f16b2e5053170d0487d : SUCCESS in 2m 24s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f6513ad2f0884abfa31b3456c7b2125c : SUCCESS in 3m 11s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/2cd73d859bb042299b485d26d210f453 : SUCCESS in 2m 48s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/f86535c4626e4fff94a9104874850ea6 : SUCCESS in 2m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/68370bf2273344708cacf53218d4cf25 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b4219df2a1bd01026ff725cc1177cf8adb9c23e1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-08-03 17:49:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c2140acdeda17fea3959a6f93303ca9f586d4ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 17:50:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20e12d6bf8f4e9bb6241eb447bb04bce80746370","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 17:58:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6068178422b4e00b64c6239103fecb2\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/2941d3cb376a4c759645544ca084c765 : SUCCESS in 3m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4bb651ba87af40fb95fb36862e5b2a6f : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"e526b6fea9358a29dd5656b0791fcab75fc5fad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 17:58:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ac752edf596455cf4698417da8643b49417649a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 18:00:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90180f91daa4d5d8dc4f9a3838e5938\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/5b7cc068c0594ce188b8aabf4c1cf22a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25215327e044721cbc85a8bf123a7552621cd6a1","revisions":{"25215327e044721cbc85a8bf123a7552621cd6a1":{"kind":"REWORK","_number":1,"created":"2023-08-03 15:25:47.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/32/890432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/32/890432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/32/890432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/32/890432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/32/890432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/32/890432/1"}}},"commit":{"parents":[{"commit":"ab0fad7f5878961884394a4c552256947a106fdc","subject":"infra-deb-package-build: run mk-build-deps in separate directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ab0fad7f5878961884394a4c552256947a106fdc"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2023-08-03 15:24:44.000000000","tz":60},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2023-08-03 15:24:44.000000000","tz":60},"subject":"openstack-tox-snap-with-sudo: increase timeout","message":"openstack-tox-snap-with-sudo: increase timeout\n\nIncrease the timeout for building a snap to 60 minutes; with some\nof the transient changes in the Python dependency chain the snaps\nhave to build quite a few more native extensions, pushing out the\nbuild time.\n\nChange-Id: Icd7f2e10522c2cf69107db74823bf8b767a9e53d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/25215327e044721cbc85a8bf123a7552621cd6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/25215327e044721cbc85a8bf123a7552621cd6a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
