)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~899078","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I52801caecab3db55b80e041b778a5a6cbaea9af8","project":"openstack/openstack-zuul-jobs","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-25 15:35:21.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-10-25 15:50:20.000000000","reason":"Change was submitted"},"11805":{"account":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"last_update":"2023-10-25 15:50:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I52801caecab3db55b80e041b778a5a6cbaea9af8","subject":"Update openstack-python3-charm-jobs py3 branches","status":"MERGED","created":"2023-10-23 15:48:06.000000000","updated":"2023-10-25 16:01:11.000000000","submitted":"2023-10-25 15:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899078","meta_rev_id":"b5e3d0d4066ac6c0742f2982bc11c090a30b7518","_number":899078,"virtual_id_number":899078,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-25 15:50:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-10-25 15:42:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2023-10-25 15:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2023-10-25 15:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-10-25 15:42:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-23 15:48:24.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-23 15:48:48.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-10-23 15:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-23 20:00:56.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-10-25 15:42:50.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2023-10-25 15:44:52.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4f8dfcf1437edd16d071418a0e3a51357c85ba9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-23 15:48:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53aaf4aa22d3e2cb08cdaa6413eec60f47306083","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-23 15:48:24.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: openstack-tox\n      abstract: true\n      parent: tox-fips\n      nodeset: ubuntu-jammy\n      description: |\n        Base job to run tox jobs with OpenStack project specific updates.\n  \n        Processing upper-constraints files from openstack/requirements.\n      branches:\n        regex: ^stable/(stein|train|ussuri|victoria|wallaby|xena|yoga|zed)$\n        negate: true\n      required-projects:\n        - name: openstack/requirements\n      vars:\n        tox_constraints_file: \u0027{{ ansible_user_dir }}/src/opendev.org/openstack/requirements/upper-constraints.txt\u0027\n        enable_fips: false\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6888111e141422ce172282893a7e9c0cb5258385","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-23 15:48:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dbcbfcdfe48d2919084e5ee4b0526e176408ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:59:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/493af8ffe4d44c80a1d48d163c4cd338\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0c9e0a0044bb47fead183aa2f88ed478 : SUCCESS in 2m 56s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d9ff44636b964107ba5f77523478167f : SUCCESS in 2m 57s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/cbb63b219e724c6d9d00bab7443765c8 : SUCCESS in 2m 31s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/aebf09052cdd483492e512bb1fef4593 : SUCCESS in 1m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/33ab84dbb65241bea79ae68f8a7b1611 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"26999de239c8013dd6dda25efaab93899b778832","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-10-23 17:46:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"15151f767782f9c43aec751671ff26f81c9ceacc","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-23 19:29:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be5b18bf5baf89ec7967b544c0b885e60a9d7c9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-23 19:30:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b547fe4418411b409a7d91874bd65f88a3362694","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-23 19:30:59.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: openstack-tox\n      abstract: true\n      parent: tox-fips\n      nodeset: ubuntu-jammy\n      description: |\n        Base job to run tox jobs with OpenStack project specific updates.\n  \n        Processing upper-constraints files from openstack/requirements.\n      branches:\n        regex: ^stable/(stein|train|ussuri|victoria|wallaby|xena|yoga|zed)$\n        negate: true\n      required-projects:\n        - name: openstack/requirements\n      vars:\n        tox_constraints_file: \u0027{{ ansible_user_dir }}/src/opendev.org/openstack/requirements/upper-constraints.txt\u0027\n        enable_fips: false\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f98161d486561528284382dc04263f3ae39b3855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 19:45:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b813aeea7a8e44d6a949af0da15b61a2\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d582ada094e747499110a1aed0c2a680 : SUCCESS in 2m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d7c6b32f3a364f599f8828e24fb6dba9 : SUCCESS in 3m 03s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b7e8a6e3c3014214ad317057435aaa50 : SUCCESS in 5m 54s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/f7873fb11d2d49bc8fd111882ef5a0b1 : SUCCESS in 1m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f33b31a6637543d48b31d7ba92c52080 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9ebb8ea0e99f5f9664eb4b8d6bdac97b711e6311","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-10-23 20:00:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e49d14db6af4d42f6251d35071e54e87e4bb81e5","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-24 17:20:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e21ebefe9064a48f31f8efede5296e9c26da852","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-25 14:56:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33913dda936f1c230f2c91dc4209e71c9cc385a1","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-25 15:28:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb37bb0299a514be0c89f5e9f68fc312b7019a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-25 15:30:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2d16e594d3d84056e368f673ef93b911201d142d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-25 15:30:28.000000000","message":"Patch Set 3:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: openstack-tox\n      abstract: true\n      parent: tox-fips\n      nodeset: ubuntu-jammy\n      description: |\n        Base job to run tox jobs with OpenStack project specific updates.\n  \n        Processing upper-constraints files from openstack/requirements.\n      branches:\n        regex: ^stable/(stein|train|ussuri|victoria|wallaby|xena|yoga|zed)$\n        negate: true\n      required-projects:\n        - name: openstack/requirements\n      vars:\n        tox_constraints_file: \u0027{{ ansible_user_dir }}/src/opendev.org/openstack/requirements/upper-constraints.txt\u0027\n        enable_fips: false\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":3},{"id":"0739763c4ab7e592ca79a8cbe511b665446f2892","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-25 15:35:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"46a01e5c6e91c626b81f0a9ebc71b7b1ce7945b2","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2023-10-25 15:36:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c9d3421c7785324f466eb4c55a48e5d7da32bbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 15:41:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/605ad788f6c8412da739accbda17b2fb\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/cf64fb22651d417bbcbcfdf3e40200f4 : SUCCESS in 2m 27s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8969ed9de0404decaa98d0f85a8e2d96 : SUCCESS in 2m 48s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/2f3241c3732346db9a8b6f2e2e75a3f7 : SUCCESS in 2m 19s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/9e5a333c57c445a493eb13e085cbbbc5 : SUCCESS in 1m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea8e43a702684f6a8e6d24f80ae3b9b9 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"260dc00236e9b77d49d8ca18d19f72732968e02e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-10-25 15:42:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"36d50fb118dc1bdef1c496f77b7ce781722d0809","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 15:43:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"10020721533173c7541eeb00b37b9b56e00e7f71","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-25 15:44:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e32e3205ecf8732a85ec2266a8cca8638a6906e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 15:50:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e9a531a4da843fba4f6b6e4982989c7\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/39007f2cfac64605b2fb169ad436440a : SUCCESS in 2m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca9fbbc68c984598af4fd74c67a2d064 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":3},{"id":"3d96c51c691820845b699a1bc4dfd3140e31832a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 15:50:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b5e3d0d4066ac6c0742f2982bc11c090a30b7518","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 16:01:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c69c5a7f6a314ae1b57d645804840066\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/be15ca58ae514b78afe0b89148be1c87 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6440c5a36081d16792b9892c7064f20807a477be","revisions":{"e6a7919414480ac0f5a263ff571af18277e53d50":{"kind":"REWORK","_number":1,"created":"2023-10-23 15:48:06.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/78/899078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/899078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/1"}}},"commit":{"parents":[{"commit":"bd597347e8312bcffe0f6b616b84b0e5294aee62","subject":"Prepare the job template for 2024.1 testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bd597347e8312bcffe0f6b616b84b0e5294aee62"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-23 15:41:03.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-23 15:48:05.000000000","tz":-240},"subject":"Update openstack-python3-charm-jobs py3 branches","message":"Update openstack-python3-charm-jobs py3 branches\n\nEnable py311 unit tests for master branch only and run py38/py310\nfor stable/2023.1 and stable/2023.2 branches.\n\nWe found that we have an issue running py311 tests on debian,\nand have decided to fix it in master branches only.\n\nChange-Id: I52801caecab3db55b80e041b778a5a6cbaea9af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/e6a7919414480ac0f5a263ff571af18277e53d50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/e6a7919414480ac0f5a263ff571af18277e53d50"}]},"branch":"refs/heads/master"},"90243a10598fec735e8593b3ffdfb7028c3a2938":{"kind":"REWORK","_number":2,"created":"2023-10-23 19:30:41.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/78/899078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/899078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/2"}}},"commit":{"parents":[{"commit":"bd597347e8312bcffe0f6b616b84b0e5294aee62","subject":"Prepare the job template for 2024.1 testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bd597347e8312bcffe0f6b616b84b0e5294aee62"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-23 15:41:03.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-23 19:30:38.000000000","tz":-240},"subject":"Update openstack-python3-charm-jobs py3 branches","message":"Update openstack-python3-charm-jobs py3 branches\n\nEnable py311 unit tests for master branch only and run py38/py310\nfor stable/2023.1 and stable/2023.2 branches.\n\nWe found that we have an issue running py311 tests on debian,\nand have decided to fix it in master branches only.\n\nChange-Id: I52801caecab3db55b80e041b778a5a6cbaea9af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/90243a10598fec735e8593b3ffdfb7028c3a2938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/90243a10598fec735e8593b3ffdfb7028c3a2938"}]},"branch":"refs/heads/master"},"6440c5a36081d16792b9892c7064f20807a477be":{"kind":"REWORK","_number":3,"created":"2023-10-25 15:30:09.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/78/899078/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/899078/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/899078/3"}}},"commit":{"parents":[{"commit":"bd597347e8312bcffe0f6b616b84b0e5294aee62","subject":"Prepare the job template for 2024.1 testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bd597347e8312bcffe0f6b616b84b0e5294aee62"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-23 15:41:03.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2023-10-25 15:30:06.000000000","tz":-240},"subject":"Update openstack-python3-charm-jobs py3 branches","message":"Update openstack-python3-charm-jobs py3 branches\n\nEnable py311 unit tests for master branch only and run py38/py310\nfor stable/2023.1 and stable/2023.2 branches.\n\nWe found that we have an issue running py311 tests on debian,\nand have decided to fix it in master branches only.\n\nChange-Id: I52801caecab3db55b80e041b778a5a6cbaea9af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6440c5a36081d16792b9892c7064f20807a477be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6440c5a36081d16792b9892c7064f20807a477be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
