)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~911088","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ibd09929c956371b40d5dd522020a6e39f99e9f17","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"drop-legacy","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-06 06:05:31.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"Ibd09929c956371b40d5dd522020a6e39f99e9f17","subject":"Drop lib-forward-testing template","status":"MERGED","created":"2024-03-05 11:22:05.000000000","updated":"2024-03-06 06:17:42.000000000","submitted":"2024-03-06 06:15:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911088-drop-legacy","meta_rev_id":"1fc214be1bc96fd6fab84a198f0adaa960fda93a","_number":911088,"virtual_id_number":911088,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 06:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2024-03-05 14:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2024-03-06 06:05:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-03-05 15:25:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2024-03-06 06:05:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 11:22:22.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 11:22:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 11:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-05 14:50:55.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2024-03-05 15:25:34.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"43cc305d9fe95837417c9d63535d8991d6597eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-05 11:22:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e27913ac75dbdf6d9b98a3c6fff1a5978cd20b9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-05 11:22:22.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: openstack-tox\n      abstract: true\n      parent: tox-fips\n      nodeset: ubuntu-jammy\n      description: |\n        Base job to run tox jobs with OpenStack project specific updates.\n  \n        Processing upper-constraints files from openstack/requirements.\n      branches:\n        regex: ^stable/(victoria|wallaby|xena|yoga|zed)$\n        negate: true\n      required-projects:\n        - name: openstack/requirements\n      vars:\n        tox_constraints_file: \u0027{{ ansible_user_dir }}/src/opendev.org/openstack/requirements/upper-constraints.txt\u0027\n        enable_fips: false\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":1},{"id":"303193f6ebd42ccc8eae0080f3d1bccabc1d2ac7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 11:22:26.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/pbr on branch master.  The\nproblem was:\n\n  The project template \"lib-forward-testing\" was not found.\n\nThe problem appears in the the \"opendev.org/openstack/pbr\" project stanza:\n\n  project:\n      templates:\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n      check:\n        jobs:\n          - openstack-tox-pep8\n  ...\n\n  in \"openstack/pbr/.zuul.yaml@master\", line 115","accounts_in_message":[],"_revision_number":1},{"id":"9f634d4ca8a3233e2e0a4d73ddbe4a907accc782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 11:22:34.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/pbr on branch master.  The\nproblem was:\n\n  The project template \"lib-forward-testing\" was not found.\n\nThe problem appears in the the \"opendev.org/openstack/pbr\" project stanza:\n\n  project:\n      templates:\n        - lib-forward-testing\n        - lib-forward-testing-python3\n        - periodic-stable-jobs\n        - publish-openstack-docs-pti\n      check:\n        jobs:\n          - openstack-tox-pep8\n  ...\n\n  in \"openstack/pbr/.zuul.yaml@master\", line 115","accounts_in_message":[],"_revision_number":1},{"id":"7566a0b69fc5579fc7e7922b492bf5ae13348af2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-05 11:31:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"000121b7de89fa265eff1c9e07bbddb4532f10bc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-05 13:13:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3947ca372bfe165630ea6df35499ad93e42f2e0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-05 13:13:48.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: openstack-tox\n      abstract: true\n      parent: tox-fips\n      nodeset: ubuntu-jammy\n      description: |\n        Base job to run tox jobs with OpenStack project specific updates.\n  \n        Processing upper-constraints files from openstack/requirements.\n      branches:\n        regex: ^stable/(victoria|wallaby|xena|yoga|zed)$\n        negate: true\n      required-projects:\n        - name: openstack/requirements\n      vars:\n        tox_constraints_file: \u0027{{ ansible_user_dir }}/src/opendev.org/openstack/requirements/upper-constraints.txt\u0027\n        enable_fips: false\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master\", line 17, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1125749b329e8f0385a6a71bf41b8262f3b2829c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 13:26:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26bc73c22a2b46fa8c0193c0b574b1c3\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/20145997afa84aa98baa88aefcb1e050 : SUCCESS in 3m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/26fe6e9e37f34e979b42b39010665076 : SUCCESS in 3m 48s\n- openstack-infra-extra-integration-xenial https://zuul.opendev.org/t/openstack/build/b0a960bd7cfb410da758779947b6fab0 : SUCCESS in 2m 56s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/64256548dd1848ce97a07a142ee9c41c : SUCCESS in 2m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dc2c5b48d64c4b0f86b34f70615ec28f : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ce8550fe3d86c5333c6eb4bc89850aebe4edcd57","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-03-05 14:50:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e215509ac2e200f057f5bfcaceb2ff358633eb6e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-05 15:25:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1d3856ec9c6edacfd9ea57c8c39573c50e89e930","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-06 06:05:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9cf7ddf42d1231e0d4d98181502c8bbde903e6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:05:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d0e8956a66d3d6ca0f2ab5a84f51d8dab5912f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:15:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a1e526ed0724c3ca6bade8c5263b03b\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/401b42e44813473399893fe2be78bf35 : SUCCESS in 1m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a562f85f599845d0ae76a6b3daf031e7 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"2d49ae8e92f8a2aa13ff231b7334ab3868fdf5a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:15:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1fc214be1bc96fd6fab84a198f0adaa960fda93a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 06:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed6793586f6e404fb5daac6cab099414\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/3baa0715e1c94cd6997bd86c71297e8a : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"023d6d494d89d6f2a14125331a49ecef184d7dc5","revisions":{"023d6d494d89d6f2a14125331a49ecef184d7dc5":{"kind":"REWORK","_number":1,"created":"2024-03-05 11:22:05.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/88/911088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/88/911088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/88/911088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/88/911088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/88/911088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/88/911088/1"}}},"commit":{"parents":[{"commit":"0415d86ac390783f77933527fafc9b603a976aee","subject":"Drop integrated-gate template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0415d86ac390783f77933527fafc9b603a976aee"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-05 11:20:40.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-03-05 11:20:40.000000000","tz":60},"subject":"Drop lib-forward-testing template","message":"Drop lib-forward-testing template\n\nThis uses the legacy tempest-full job which is about to be removed from\ntempest. All uses should refer to the new lib-forward-testing-python3\ntemplate instead.\n\nChange-Id: Ibd09929c956371b40d5dd522020a6e39f99e9f17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/023d6d494d89d6f2a14125331a49ecef184d7dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/023d6d494d89d6f2a14125331a49ecef184d7dc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
