)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~921878","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I2d292baa2213a659fda61724e00799bf4bff7d05","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"weblate-client","attention_set":{},"removed_from_attention_set":{"38225":{"account":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"last_update":"2025-10-14 11:01:52.000000000","reason":"\u003cGERRIT_ACCOUNT_38225\u003e replied on the change","reason_account":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"}},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2026-02-04 09:49:52.000000000","reason":"Change was abandoned"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-12-03 15:47:04.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-10-18 15:44:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"37971":{"account":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"last_update":"2026-02-04 09:49:52.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-10-20 06:54:45.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"23279":{"account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"last_update":"2026-02-04 09:49:10.000000000","reason":"\u003cGERRIT_ACCOUNT_23279\u003e replied on the change","reason_account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}}},"hashtags":[],"change_id":"I2d292baa2213a659fda61724e00799bf4bff7d05","subject":"Add ansible role for weblate client configuration","status":"ABANDONED","created":"2024-06-12 16:07:21.000000000","updated":"2026-02-04 09:49:52.000000000","total_comment_count":57,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"8a35c08b9c0265c43e698d2f169f98d10b07349a","_number":921878,"virtual_id_number":921878,"owner":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-18 14:13:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},{"value":-1,"date":"2025-12-29 22:29:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},{"value":0,"date":"2026-02-04 09:37:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-12 16:08:34.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2024-06-12 16:17:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-19 02:21:41.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2025-07-08 21:35:32.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2025-09-13 15:40:17.000000000","updated_by":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"reviewer":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"state":"REVIEWER"},{"updated":"2025-09-18 12:00:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-09-24 15:06:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2025-09-27 12:44:51.000000000","updated_by":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"reviewer":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"state":"REVIEWER"},{"updated":"2025-10-18 15:44:55.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-10-31 09:25:23.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2025-11-26 16:53:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"29d7ae247bd1818c62412a6c6cbd2f85dbcee960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2024-06-12 16:07:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a86a03a32d3d765fb3958c284656892738e3007","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2024-06-12 16:08:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"8de767bbd26c41e6b5ab3456223f0559238e62cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 16:17:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4846a3f5221c4e09a2c086f620905474\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/1c0839725c0545d485531ff8cbdd6e64 : SUCCESS in 2m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/77f8c481d95f40758bef4d5ac36f8e25 : SUCCESS in 2m 52s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/77679b7685ca46e4a6775ef6053f97da : SUCCESS in 1m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/03f01249ee594bdd922cd82d681be436 : FAILURE in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"42dcd7b7c6629b20c417138dcb3e11c58f5d6b4a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-06-12 23:29:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a6212699fc28c5330910918b9006ab2c1a98b058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 23:42:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cec4198090f648999b88afe966531880\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/27797fd4169a479ebfebff50c9465531 : SUCCESS in 2m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c3635c3477694135ae348e85baf3d7e6 : SUCCESS in 2m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/86f7fe3e2a4b4e479ec37e53bb4599bb : FAILURE in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"84d1c401e3018a740f749fc001684be93c174c8c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-06-12 23:47:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"893a2d7a148f760412044c3854ff3b0e5906f628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 23:57:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aebc2c486fd74b03963c676f4f5ff593\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6aa9615faeeb4d05884e630cfaad1779 : SUCCESS in 2m 01s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4153498a17874832a6441c97e3afbcdc : SUCCESS in 2m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ba4b3e952f254e8cb9677670c1b77982 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"5bdc3ac5a7dfee697b8f0842f41496aceb7d00a8","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-06-19 02:21:41.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7c5e7f2b3872df36e46316c752f366f31388f4c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-07-08 21:35:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"591b5df0b74d45c249f94d1d218c4fe991c598d6","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-07-08 21:37:57.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2fee5eb10aa5d97b9b2b0581649c5052d06a794c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 21:45:23.000000000","message":"Patch Set 4: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/093adbf476df4991b7fd985791daf118\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/41a77459e8724fc38cebfbcc30d0659d : SUCCESS in 2m 34s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8bc75799103c425ba88168828bdde95d : SUCCESS in 3m 06s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5a671c5e0a4f4f25800f991fbafcca3b : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"5a9d30ae781a592f912afdd20180e3c544ccb44a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-07-08 22:04:42.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"882f43a2de108253f60546768bc4ff3392f9348b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 22:09:50.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f681b5e6ad794ec4bf94d3299ae72100\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/3a856bf5bd6c48049d55550fa4f5a616 : SUCCESS in 3m 18s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ca2c1d87a05941b39071aad1a5250fb1 : SUCCESS in 2m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b8e6b70fb5bc478da04ef02883a2fb96 : FAILURE in 3m 52s","accounts_in_message":[],"_revision_number":5},{"id":"e1c95d8f67a1b6625dec2dc9a4b67c1d6db6cb6e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-07-08 22:32:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e6e5866c7d0e9e754a860241b5bdd3d161b126f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 22:49:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ecdb91fa25a49deaf5331352aa80429\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4bddeedd5b8b4b41936fdfbd39cf55b9 : SUCCESS in 3m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/687310d2ec33407f8cc87346bbe17b0b : SUCCESS in 3m 34s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9e742a5b1bb74fe1a694c486f9db5a00 : FAILURE in 1m 58s","accounts_in_message":[],"_revision_number":6},{"id":"42b8d4e8b127dc2c3c60e46649145742249c57f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-13 15:40:17.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"411234e8f7a13c012e2da4cb94ada4d3ac9effb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 15:48:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8094a1be18ea4f3a9ee9bbeca4a56052\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ca30c118b6b34f3484a3e58835125f4a : SUCCESS in 2m 21s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ee9683bcf2fe47ea8d6a9c609393c71b : SUCCESS in 3m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8b8c64ede2b14ab6b9fd195e52103a62 : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":7},{"id":"cc1b81552dafd315fae92826d7fa8386599717ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-14 14:25:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"0179e2f4c91522d676ce3eaa2dd4be711ad6d4e2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-14 14:28:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"59092d45abd2c174f33c609222a6650a635157ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-14 14:37:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a36b7ff1d79e4d7b9bfe4002131c3be9\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/8b59d601c38d4744a956207dc4749f64 : SUCCESS in 2m 13s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7bccf8a4169c468280861afbb0c3d608 : SUCCESS in 3m 07s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aab30aec259f447d95d9ce8472a5ab97 : FAILURE in 4m 09s","accounts_in_message":[],"_revision_number":9},{"id":"c0db4a1c66a526adfca405e68b8f09634d45e33d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-17 00:27:16.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"ebd8cb003ffa1d9231d5b21a97075f4662ee06e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 00:53:10.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26f1a56ec308477bbd4c15943f9a9198\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6f006f1f5a244be896ab5bcdb4178490 : SUCCESS in 3m 13s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/470b2464d9324024bfbad83f0a3f5e96 : SUCCESS in 4m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b5c4a880315e459f830a327646893426 : FAILURE in 5m 22s","accounts_in_message":[],"_revision_number":10},{"id":"d54158f3af7c04051afbefddede705f1dcf38b11","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-17 13:21:42.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"6d6bfee2e9234aea3e3e7fbb66ed04ca3db73020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 13:41:08.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/457cc22806ba48118497c6ea19d45845\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/a2bfce3ae6064ec59ca70e2eba0a92cd : SUCCESS in 2m 31s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ac3adb71fa23424f9ce4608b6c38e103 : SUCCESS in 2m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fee2ca7d4e0e48af8b7a3edee1cf8fc4 : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":11},{"id":"d913bc1b120eda410848a5f279adc355a5307534","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-18 11:27:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":11},{"id":"5de9fa4a592e51a297656a149ff4db40ba075b56","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-09-18 12:00:05.000000000","message":"Patch Set 11: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":11},{"id":"89ba1177c5546d5a6c9756ebfd0655b6deab2a27","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-18 12:53:09.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e240bf370f01201c6d6df26fae61ac544935a550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-18 12:57:49.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"528923764213c98e0afc8fbc2123a27fb514f5c1","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-18 12:58:50.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e793d593d27465d66d53b02b63f6ff066ca36a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 13:20:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e71c5e0e14e44a9c9107823f52f426ab\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/9f45afa769634a5d803120512e27e8d8 : SUCCESS in 1m 30s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c99a4a2f25714f3d9242c40546266b40 : SUCCESS in 3m 14s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4bd1e623df9d48f68d0be50f7b0d7cdd : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":12},{"id":"55c89b41e188bb438975736d0026e099fca0dd2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-18 15:00:33.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"c688f7ad5a6bd705b6518b352460af35c4a87ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 15:24:02.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b7d52a1f23d46ed955eacf1af939676\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/31a36a69febd4be399213fec5464e082 : SUCCESS in 2m 10s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/15a0aea5f60542e285ec36c501792ab8 : SUCCESS in 3m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cfc67816c8844ee7b783db0874c18d81 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":13},{"id":"5871597c543ad376bf8c7a124a94d2f18e9aead0","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-20 15:51:40.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"135ed57147a5cc1c5cb86e527c67256e7ac64801","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-09-24 15:06:08.000000000","message":"Patch Set 13:\n\n(9 comments)","accounts_in_message":[],"_revision_number":13},{"id":"7de5ca1fc5d81ee0d02157fe7c12f01bd9f8ff0a","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-24 15:28:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"6f1d1cde076676bbdac8e2743b69818d97478e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"date":"2025-09-27 12:44:51.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"a2188b7907dfb9a290625c61368307b26899c6bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 12:52:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b097c39267747ec8fa94415b2a74cc1\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d63c92644c5842849de09766ad8cae39 : SUCCESS in 2m 20s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cb345e1055ca4315ac9f776552795836 : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/340534e1875949b8833d8211ec45720a : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":14},{"id":"1f5b1f5a7bf8054431df8679fa0803841f5698e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-27 13:07:40.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"cc6d0517965b6c97821314e94d972a2e4540c472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 13:15:49.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c34a1aefb9247299503ada1016fac8f\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/83847995a6ea487d90b326338640af7b : SUCCESS in 3m 12s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bc1a254414a24506b90a22e8c85ca630 : SUCCESS in 3m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3276d7c0d5b64b1f84aa3abc79acf020 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":15},{"id":"5ccc23430822d1ec17de9e881482f1590eb8c031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-27 13:17:40.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"e8a9b0f762a8b23bccaa37ada5aac8e5b34079a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 13:28:21.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3d8661513fe43b0b08e660e02b97b57\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d0a150dbd9b14564b7cf6cd62183fed8 : SUCCESS in 2m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/71b3be71350a41909d1bd3084e2a13ba : SUCCESS in 2m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b42ae9b326484544b4895e2799bb08df : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":16},{"id":"ea13cb5bfd6818de670363998bfd41f4c28c19be","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-09-27 14:28:40.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"423ae28b37b8515943fc7363bbdf199805ad3fcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-09-27 16:20:55.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"5091b4150121b836fbef3d58b714926278c84dfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-27 16:34:35.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba31acb451643cf9a6a380963f026e3\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/9a3b3891f54041229552cac490b55f61 : SUCCESS in 4m 15s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2b52dab04f774ac088df2059310e1625 : SUCCESS in 3m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/12f26b14d2ce449dbcd73b17b328efb3 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":17},{"id":"217717fb3562118aa809e32527ea036a19d24d58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"date":"2025-09-28 15:36:01.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"b18291d4d60146b0c92427c4faeae175337fd89b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-28 15:45:37.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a38d83110dba446c95374010bc73f693\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/9b2ab8563a7d475a98de7d1abafd985b : SUCCESS in 3m 22s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d98bd64c95ec42c3a7e44693b5fbaf79 : SUCCESS in 2m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/17af9b8e57d24a8594c6cb7d6d471566 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":18},{"id":"081c378697b7d83426f6b6bef9f7c838d95f7f97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-10-01 10:39:51.000000000","message":"Patch Set 19: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"9c0d7c8819bc0168baad6c2f0d028d831940a758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 10:48:44.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cde6ad71b0d4783bcc31ae5e799aa45\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/27a088efc2c443e0a7ce274fca812a45 : SUCCESS in 1m 32s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a6ef92755e2449c1b4da045ba72a134a : SUCCESS in 3m 09s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a1da48a17ecc47e69290773b9508f365 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":19},{"id":"4bd3f4e48347808b4a02266a42ba3c0ffd7b6496","author":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"date":"2025-10-14 11:01:52.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"dc497d91ad369fdd2db12ad663059bfd3a399433","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-10-14 13:15:54.000000000","message":"Patch Set 19:\n\n(8 comments)","accounts_in_message":[],"_revision_number":19},{"id":"60560cf95fe60ded69f698cc8ee02e17640bf1db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-10-14 17:03:24.000000000","message":"Patch Set 20: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"02d91a493912e81c8614bfe055d069040a28dd0c","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-10-14 17:03:33.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"27c8d0f19f5d6778fb56f21534f0d76c92ff045f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 17:13:01.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0360a912ccf49d187703d68cb92b541\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d718672bbd40420997f5924c37181c56 : SUCCESS in 2m 30s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/14422100b09147f2878e51727490ba31 : SUCCESS in 2m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3ddd8d23fce54f9bbf9de793f2933ab9 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":20},{"id":"02cd4e8cda5f28e25dbc5f9935fdf75cdd62cd16","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-10-17 11:48:31.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2660a4d388790a7ba9a6dc90112c7a34b6156d31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-10-18 14:10:06.000000000","message":"Patch Set 21: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"0f2e89356daf4facc1a6176957001e558f4cd6cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-18 14:13:33.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7bf2df8616a4dcc81082341a59adf72\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/736555134e9d4b9890190fab75d70faf : SUCCESS in 2m 01s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/711cf35154414043baed51d75029bf88 : SUCCESS in 2m 59s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/478278b77ac045cab47f78e662220acc : SUCCESS in 1m 58s","accounts_in_message":[],"_revision_number":21},{"id":"b17617ab94b7eae9309251be1e4cb4b8122dfb62","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-10-20 06:54:45.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"6ec9ad02b5e479b5d7614407aab6caa053fffda5","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-10-28 22:37:11.000000000","message":"Patch Set 21:\n\n(6 comments)","accounts_in_message":[],"_revision_number":21},{"id":"dcf07b7b096455d5cfeccd3dbd4f72ae09e0e8df","author":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"date":"2025-11-01 04:36:59.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2e8198ae519e76aeb11397c41283a1aa9739d84a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-12-03 15:47:04.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"081a871e94d676ca4d512bdd2cf999a3d8e58c36","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-12-29 22:29:17.000000000","message":"Patch Set 21: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"ed46f00710bd9717e2fbbe4fd94ad229373142b9","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-12-30 21:39:38.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"6991d2488639f6f89918179714fa6f4aa4f7d2e9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-02-04 09:37:18.000000000","message":"Patch Set 21: -Code-Review","accounts_in_message":[],"_revision_number":21},{"id":"805ad25c730ce40239ddad4c5d7f594a9f1e7f71","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2026-02-04 09:49:10.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"8a35c08b9c0265c43e698d2f169f98d10b07349a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2026-02-04 09:49:52.000000000","message":"Abandoned\n\nI will submit the patch again after the i18n team reviews the scripts first.","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"b9dac8fed0326423358baf0a78b7ac80fc4f28ff","revisions":{"1f70c4414590690c627c07e5746a6638405cb42b":{"kind":"REWORK","_number":1,"created":"2024-06-12 16:07:21.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/78/921878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/1"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:07:02.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will be replace the previously used Zanata client\nconfiguration.\n\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1f70c4414590690c627c07e5746a6638405cb42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1f70c4414590690c627c07e5746a6638405cb42b"}]},"branch":"refs/heads/master"},"6bd863890ea6d2f3a18eb2baff5f848c8a1277f8":{"kind":"REWORK","_number":2,"created":"2024-06-12 23:29:42.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/78/921878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/2"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-06-12 23:29:38.000000000","tz":0},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will be replace the previously used Zanata client\nconfiguration.\n\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6bd863890ea6d2f3a18eb2baff5f848c8a1277f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6bd863890ea6d2f3a18eb2baff5f848c8a1277f8"}]},"branch":"refs/heads/master"},"356671fdbf1e592765f6ca38f47d26ba7c515c94":{"kind":"REWORK","_number":3,"created":"2024-06-12 23:47:53.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/78/921878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/3"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2024-06-12 23:47:53.000000000","tz":0},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will be replace the previously used Zanata client\nconfiguration.\n\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/356671fdbf1e592765f6ca38f47d26ba7c515c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/356671fdbf1e592765f6ca38f47d26ba7c515c94"}]},"branch":"refs/heads/master"},"bb5abdf89bacd1b939e029a153ea52e8a2b9f854":{"kind":"REWORK","_number":4,"created":"2025-07-08 21:35:32.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/78/921878/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/4"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-07-08 21:35:14.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bb5abdf89bacd1b939e029a153ea52e8a2b9f854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/bb5abdf89bacd1b939e029a153ea52e8a2b9f854"}]},"branch":"refs/heads/master"},"3837b8d755e297f987c1e092647ba15cd4c3fea9":{"kind":"REWORK","_number":5,"created":"2025-07-08 22:04:42.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/78/921878/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/5"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-07-08 22:04:33.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3837b8d755e297f987c1e092647ba15cd4c3fea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3837b8d755e297f987c1e092647ba15cd4c3fea9"}]},"branch":"refs/heads/master"},"f268689e3cd259eff9fddc005049a4c162cef1a8":{"kind":"REWORK","_number":6,"created":"2025-07-08 22:32:37.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/78/921878/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/6"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-07-08 22:32:27.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f268689e3cd259eff9fddc005049a4c162cef1a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f268689e3cd259eff9fddc005049a4c162cef1a8"}]},"branch":"refs/heads/master"},"32e301b90775b6bd95f61e39e23b9c4abe7d153a":{"kind":"REWORK","_number":7,"created":"2025-09-13 15:40:17.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/7"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-13 15:38:37.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/32e301b90775b6bd95f61e39e23b9c4abe7d153a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/32e301b90775b6bd95f61e39e23b9c4abe7d153a"}]},"branch":"refs/heads/master"},"1849ab6d068542b43e914c5c47f7d52df7a38db7":{"kind":"REWORK","_number":8,"created":"2025-09-14 14:25:17.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/8"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-14 14:24:59.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1849ab6d068542b43e914c5c47f7d52df7a38db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1849ab6d068542b43e914c5c47f7d52df7a38db7"}]},"branch":"refs/heads/master"},"ca08c52f6c4010b8bc8ccad0c28b72fb480e1aa6":{"kind":"REWORK","_number":9,"created":"2025-09-14 14:28:31.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/9"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-14 14:28:18.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ca08c52f6c4010b8bc8ccad0c28b72fb480e1aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ca08c52f6c4010b8bc8ccad0c28b72fb480e1aa6"}]},"branch":"refs/heads/master"},"583e2da29cb4ef1857e234825ff060f27d5f367c":{"kind":"REWORK","_number":10,"created":"2025-09-17 00:27:16.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/10"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-17 00:26:37.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/583e2da29cb4ef1857e234825ff060f27d5f367c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/583e2da29cb4ef1857e234825ff060f27d5f367c"}]},"branch":"refs/heads/master"},"99f740f0687e1f8b218f7efa24bbd4383b04a565":{"kind":"REWORK","_number":11,"created":"2025-09-17 13:21:42.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/11"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-17 13:21:27.000000000","tz":540},"subject":"[WIP] Add ansible play for weblate client configuration","message":"[WIP] Add ansible play for weblate client configuration\n\nAdded a new Ansible playbook to configure the Weblate client.\nThis patch will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/99f740f0687e1f8b218f7efa24bbd4383b04a565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/99f740f0687e1f8b218f7efa24bbd4383b04a565"}]},"branch":"refs/heads/master"},"4a64619354aeecc8c7a600bd2f9bff379c4e5c06":{"kind":"REWORK","_number":12,"created":"2025-09-18 12:57:49.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/78/921878/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/12"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2025-09-18 12:57:39.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/4a64619354aeecc8c7a600bd2f9bff379c4e5c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/4a64619354aeecc8c7a600bd2f9bff379c4e5c06"}]},"branch":"refs/heads/master"},"8c4c0abde2d9c48f197668535527ca99122fe6ce":{"kind":"REWORK","_number":13,"created":"2025-09-18 15:00:33.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/13"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-18 15:00:09.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0abde2d9c48f197668535527ca99122fe6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0abde2d9c48f197668535527ca99122fe6ce"}]},"branch":"refs/heads/master"},"ba1f03f753aab422f4e7655e0a2d37189d6444d4":{"kind":"REWORK","_number":14,"created":"2025-09-27 12:44:51.000000000","uploader":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"ref":"refs/changes/78/921878/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/14"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"kamothi","email":"rhkddlf7911@naver.com","date":"2025-09-27 12:44:48.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ba1f03f753aab422f4e7655e0a2d37189d6444d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/ba1f03f753aab422f4e7655e0a2d37189d6444d4"}]},"branch":"refs/heads/master"},"2163a8d66f83d3f5700bda8f1c0307dd77e41614":{"kind":"REWORK","_number":15,"created":"2025-09-27 13:07:40.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/15"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-27 13:03:56.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2163a8d66f83d3f5700bda8f1c0307dd77e41614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2163a8d66f83d3f5700bda8f1c0307dd77e41614"}]},"branch":"refs/heads/master"},"a23d6f259b82cbd0721907d49759e1efa9bf4ff2":{"kind":"REWORK","_number":16,"created":"2025-09-27 13:17:40.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/16"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-27 13:17:24.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a23d6f259b82cbd0721907d49759e1efa9bf4ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a23d6f259b82cbd0721907d49759e1efa9bf4ff2"}]},"branch":"refs/heads/master"},"6eaa55bc6bc16f1f359922c73d2599648ab2ae29":{"kind":"REWORK","_number":17,"created":"2025-09-27 16:20:55.000000000","uploader":{"_account_id":37971,"name":"Chaemin Lim","display_name":"antraxmin","email":"antraxmin@naver.com","username":"antraxmin"},"ref":"refs/changes/78/921878/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/17"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Antraxmin","email":"antraxmin@naver.com","date":"2025-09-27 16:20:26.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6eaa55bc6bc16f1f359922c73d2599648ab2ae29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6eaa55bc6bc16f1f359922c73d2599648ab2ae29"}]},"branch":"refs/heads/master"},"405a9eed6ee84bbbf2bd8d8d9bbad78ef94c74b2":{"kind":"REWORK","_number":18,"created":"2025-09-28 15:36:01.000000000","uploader":{"_account_id":38225,"name":"KIM KWANG IL","display_name":"kwangil","email":"rhkddlf7911@naver.com","username":"kwangil"},"ref":"refs/changes/78/921878/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/18"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"kamothi","email":"rhkddlf7911@naver.com","date":"2025-09-28 15:35:46.000000000","tz":540},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/405a9eed6ee84bbbf2bd8d8d9bbad78ef94c74b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/405a9eed6ee84bbbf2bd8d8d9bbad78ef94c74b2"}]},"branch":"refs/heads/master"},"418846c32ac96a6ad5c5b455a378cc0dd53163ec":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2025-10-01 10:39:51.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/78/921878/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/19"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2025-10-01 10:39:51.000000000","tz":0},"subject":"Add ansible play for weblate client configuration","message":"Add ansible play for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/418846c32ac96a6ad5c5b455a378cc0dd53163ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/418846c32ac96a6ad5c5b455a378cc0dd53163ec"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5a4fa3934196e4e94d211fcb1adab74334955cf0":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2025-10-14 17:03:24.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/78/921878/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/20"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2025-10-14 17:03:24.000000000","tz":0},"subject":"Add ansible role for weblate client configuration","message":"Add ansible role for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5a4fa3934196e4e94d211fcb1adab74334955cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/5a4fa3934196e4e94d211fcb1adab74334955cf0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b9dac8fed0326423358baf0a78b7ac80fc4f28ff":{"kind":"NO_CODE_CHANGE","_number":21,"created":"2025-10-18 14:10:06.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/78/921878/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/78/921878/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/78/921878/21"}}},"commit":{"parents":[{"commit":"8c4c0e5aa32522e7823a0efde2116738d9d8a772","subject":"Drop devstack-gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8c4c0e5aa32522e7823a0efde2116738d9d8a772"}]}],"author":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2024-06-12 16:06:53.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2025-10-18 14:10:06.000000000","tz":0},"subject":"Add ansible role for weblate client configuration","message":"Add ansible role for weblate client configuration\n\nAdded a new Ansible role to configure the Weblate client.\nThis role will replace the previously used Zanata client\nconfiguration.\n\nCo-authored-by: Antraxmin \u003cantraxmin@naver.com\u003e\nCo-authored-by: kamothi \u003crhkddlf7911@naver.com\u003e\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Antraxmin \u003cantraxmin@naver.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\nSigned-off-by: kamothi \u003crhkddlf7911@naver.com\u003e\nChange-Id: I2d292baa2213a659fda61724e00799bf4bff7d05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b9dac8fed0326423358baf0a78b7ac80fc4f28ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b9dac8fed0326423358baf0a78b7ac80fc4f28ff"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
