)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~926879","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ic96a77e4b8ae0fd5ec72de5a06cafba391509041","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"pin-reno-jammy","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-22 07:54:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic96a77e4b8ae0fd5ec72de5a06cafba391509041","subject":"Pin build-openstack-releasenotes job to jammy","status":"MERGED","created":"2024-08-22 07:22:06.000000000","updated":"2024-08-22 08:04:53.000000000","submitted":"2024-08-22 07:54:28.000000000","submitter":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926879-pin-reno-jammy","meta_rev_id":"df3d9d3dde8193de4537b62183cf5606c3a9c928","_number":926879,"virtual_id_number":926879,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-22 07:34:18.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-22 07:54:28.000000000","_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-22 07:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-22 07:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 07:34:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 07:54:28.000000000","updated_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"reviewer":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"state":"REVIEWER"}],"messages":[{"id":"d7c8467b035912404e1d962711e023c0a8bec9b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-22 07:22:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83e6534ff3fbd0b91959dedd1e2de5e56288e90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 07:34:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e37c7a7634ad4ac7a93ad0a893858c94\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/03bc33fb6dc346d1aaa90ab2ff95186d : SUCCESS in 2m 38s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/558a1e23ec83494e8c62eac69cddf2ce : SUCCESS in 2m 53s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/4f8ee0652ffd4651845eaa33df7dc111 : SUCCESS in 2m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f6487b2010314698ad5d7be0bcd5f4c6 : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0745de585dba4f7ef391439bf30fbd990686002d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-22 07:53:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2d7a3fd10816788a84a614cfe7ed2c81ab13639","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-08-22 07:54:28.000000000","message":"Patch Set 1: Verified+2\n\nBypassing Zuul to merge this and unblock openstack CI","accounts_in_message":[],"_revision_number":1},{"id":"03d5982811494ea2ebaffe5b39f45949a5d773f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-08-22 07:54:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4ca4a028ead7a1bcbe07ebbb5113606d90b325cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 07:56:36.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07bd40ed01814ab09a7d57fe81d6487b\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/47e76090e20b4e768ae58933f0f921b4 : FAILURE in 31s","accounts_in_message":[],"_revision_number":1},{"id":"df3d9d3dde8193de4537b62183cf5606c3a9c928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 08:04:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc67d9d6c94e4bc198971465735ecfd9\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/ba912f9811684ed6884a3cc1a3fb422f : SUCCESS in 2m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0eea0d7d97d94b61bbde396a15784ed7 : SUCCESS in 2m 48s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/1406bec76d094d1fba0b89a855abda70 : SUCCESS in 1m 37s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/df34ba8f29f04053b4073ca07226295b : FAILURE in 5m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8857de3fe010ccbe46181b3f56d2a455a40f1cfe","revisions":{"8857de3fe010ccbe46181b3f56d2a455a40f1cfe":{"kind":"REWORK","_number":1,"created":"2024-08-22 07:22:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/79/926879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/79/926879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/926879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/926879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/926879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/79/926879/1"}}},"commit":{"parents":[{"commit":"824b30fd1b6dc1764f8a4fd37ec57b8f1a388d81","subject":"Merge \"Prepare python testing templates for stable/2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/824b30fd1b6dc1764f8a4fd37ec57b8f1a388d81"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-22 07:20:36.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-22 07:20:36.000000000","tz":120},"subject":"Pin build-openstack-releasenotes job to jammy","message":"Pin build-openstack-releasenotes job to jammy\n\nThere are failures with building the pcre wheel on noble, we don\u0027t want\nto amend bindep for every repo, so let\u0027s run this job on jammy for new\nand try to get rid of the pcre library then.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: Ic96a77e4b8ae0fd5ec72de5a06cafba391509041\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8857de3fe010ccbe46181b3f56d2a455a40f1cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/8857de3fe010ccbe46181b3f56d2a455a40f1cfe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
