)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~930590","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I024ca7239b1e8a1a5c9aaa72576b0e8a891116fc","project":"openstack/openstack-zuul-jobs","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-09-27 16:00:51.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2026-07-10 09:26:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I024ca7239b1e8a1a5c9aaa72576b0e8a891116fc","subject":"Build wheel cache for Ubuntu Noble","status":"ABANDONED","created":"2024-09-26 14:40:57.000000000","updated":"2026-07-10 09:26:24.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f8bf62351fb3afc13e0d2d68064a7ecf2a7e7a18","_number":930590,"virtual_id_number":930590,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-26 15:58:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-26 14:50:29.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2024-09-26 15:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-26 15:58:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55fe0633a9fbc67ef6f51ece0e24a2a414fb4c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-09-26 14:40:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff24b13424b01612594c3b6b48c1aed917df74bc","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-09-26 14:50:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef8c90c3846c7535a944be8384750ec04cd7d775","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-09-26 14:54:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b2dea6802f0095b2fd3d5c46792c9522dad2094","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 15:35:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d55e68af9bb42668f02228024997ea7\n\n- build-wheel-cache-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/ee1a0f0f0f524efda5d6f56c4a5007f5 : SUCCESS in 40m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4d41e5a6ea56077dee7e10e4b5042bf33cd7487a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-09-26 15:56:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bccccfd540b672cedd4d3a177bc69db19858a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-26 15:58:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264473553cf348bfa8f248e4766b5f4b\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bdd8f3c1c5954562860dec02094812f5 : SUCCESS in 4m 55s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2afbd61533ff45dfb1f3ca01e9209cb1 : SUCCESS in 4m 43s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/d76f13aeb2974736ba3a088d5276c1c1 : SUCCESS in 2m 12s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63ebda10e9824fb99104d8c26efa6276 : SUCCESS in 5m 09s\n- build-wheel-cache-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6855e1bd4bb4692a84c296949f87bc7 : SUCCESS in 1h 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6613fa46f1bf27fb0dcc7b8bb4a3168aee028801","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-09-27 13:35:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"011e92f216f1247d5e8e48ca22ed904fd94bcdf8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-09-27 16:00:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8bf62351fb3afc13e0d2d68064a7ecf2a7e7a18","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-07-10 09:26:24.000000000","message":"Abandoned\n\nSeems we are doing fine without these, let\u0027s consider the wheel caches deprecated, we will clean them up once the distros that still use them go EOL","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9981adb0da167fa8facee76aff3978232c0b8d8","revisions":{"b9981adb0da167fa8facee76aff3978232c0b8d8":{"kind":"REWORK","_number":1,"created":"2024-09-26 14:40:57.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/90/930590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/90/930590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/90/930590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/90/930590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/90/930590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/90/930590/1"}}},"commit":{"parents":[{"commit":"1ce8fa37a2b54dafc24fe8a1ea5beacbe27a3ac5","subject":"Merge \"Create py312 jobs for neutron\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/1ce8fa37a2b54dafc24fe8a1ea5beacbe27a3ac5"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-09-26 14:40:36.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-09-26 14:40:56.000000000","tz":-240},"subject":"Build wheel cache for Ubuntu Noble","message":"Build wheel cache for Ubuntu Noble\n\nChange-Id: I024ca7239b1e8a1a5c9aaa72576b0e8a891116fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b9981adb0da167fa8facee76aff3978232c0b8d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b9981adb0da167fa8facee76aff3978232c0b8d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
