)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~935813","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Idb42386071d6652312bb343923473a62c55e2828","project":"openstack/openstack-zuul-jobs","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-11-22 19:54:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb42386071d6652312bb343923473a62c55e2828","subject":"Update system packages and reboot when building centos openafs","status":"MERGED","created":"2024-11-20 17:12:42.000000000","updated":"2024-11-22 19:57:48.000000000","submitted":"2024-11-22 19:54:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935813","meta_rev_id":"34ac6bcdc426d033368e97f4890d4ca1c6e0d576","_number":935813,"virtual_id_number":935813,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-22 19:54:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-22 19:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-22 19:32:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 17:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 19:32:39.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"3ac1608f6abceef15f4c14c1e8adbb3c221d0b01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-20 17:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"569fd7532fc3c7008943729fee79f6b345db1756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-20 17:18:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7cd1f64297a421ce702a872a810cd3edee4da6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 17:54:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd5f5e7574194060bdd3db5f429d4b09\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/cebce07752404c68a262e61594a92f23 : SUCCESS in 3m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/57a4446dddd24f03aea582b7643512f8 : SUCCESS in 4m 22s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/14b5db58b8c14fb18cf7933f61819e56 : FAILURE in 3m 15s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/0a5188f5db5c435eb2dcc4039c29ab5a : FAILURE in 8m 48s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/c04417a6cf1f4bbabbb1f15088659d66 : SUCCESS in 2m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5c87113f0029480aa4bb4a1df7ccaaba : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"7c0cf26379fc74b04c1a2e21c2316299e5a29c75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 00:54:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"eb8a8c8a111c174dcdff918f08d09a3a53c0e03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 01:07:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/692ae722510e4ed1b7dfa2eaa0c05f61\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/77fd7da0bd0f470b8ad9191a41f448de : SUCCESS in 2m 29s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c62504e86f424e98ae9276ed4d32d10f : SUCCESS in 3m 05s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/6ab570e22c3e49f5af9a5b9307c7051f : SUCCESS in 12m 38s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/cae9baa261c94da99c58c1cd8fc887b7 : FAILURE in 11m 03s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/d135011f37cb41af87dcb8a8f712f3a2 : SUCCESS in 1m 37s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/66a7d68f01dc4f7198838a5708bb1c7b : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"d9902ac148f19a22d7bf1bebb86250d40906e24a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 17:38:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47c2e495488ea83a3c38a23c90483a97a9c47eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 17:54:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0474a280e5845df8bcf51f33276d824\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/cf3cfd0b58ff4683a973d9cb6ce5e848 : SUCCESS in 2m 38s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/03d236cc78ce427c82ea5daa853bfb13 : SUCCESS in 3m 04s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/254a7958db8f4d60a3a7e6728eb08c92 : SUCCESS in 10m 01s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/253f030a11f4420da90f81ace664ba3c : FAILURE in 10m 31s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/deec5313a9eb40c3afb3bf6fee617af6 : SUCCESS in 1m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/891b2f8f3b5543498a0bb7c473882c27 : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"5d6d54ccf9e96a5c2ad14379b88a10526e26e8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 18:04:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f25ab5eefa8d0f85c86e95ed7e9544c5d67c9f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 18:13:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ec22f485a6e18ca3920e6424120acac86aba1cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 18:39:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7018ed0b4504ed8839fde7ef2b98363\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b5ac9316ab3846dc94804df70db70159 : SUCCESS in 2m 17s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/20e443db62344538ae4c1a1c84811204 : SUCCESS in 3m 04s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/2f97721d42f3448e961500d1391ca581 : FAILURE in 3m 14s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/2abe1eabbbfd4fa3b5f3caafcc7c0e88 : FAILURE in 8m 55s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/cedcf7145d9e406c91f32e9149a4caca : SUCCESS in 1m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2a7e5beca771477a82475723e4fd6931 : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":5},{"id":"cf78c6444ad0f1bfb0c3de624f45c12ac98e05d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 19:02:09.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ff8c51bbb6099f6d11ab081932cd6508d9740f38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 19:12:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6d4d16b59eef22d6bb24831fcd9a22cc040cd1be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 19:19:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3760d9a67a902dbab4e8f46f4b89996c2c0f8762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 19:51:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eadb30e55d040e1beb672caccb9d2c1\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/570f2fe55af64351bd4b0e3f19dfbaa6 : SUCCESS in 3m 12s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ab241113d63647499cf8d9a30b29447d : SUCCESS in 4m 17s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/ceac3784a8b24dcebead5c92ce1f84ef : SUCCESS in 15m 27s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/34d9496d091046c8a317d06f8ebfb812 : FAILURE in 9m 37s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/3f641af3a17743afa0c82ff5e357b86b : SUCCESS in 1m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/966ec02c93fc4e9082cea300baf78308 : FAILURE in 4m 18s","accounts_in_message":[],"_revision_number":8},{"id":"68872447adddb05079b20a6be165861fd7b474cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 21:59:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"fd423caaae38b33e6a665cebfe474717ab082665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:11:56.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d8065598fe44eb8857ca4eec1e20ae3\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/6a089eb451ec4bf197e8db5595a76036 : SUCCESS in 1m 50s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f70b2349a7ca4ccc8b7b1c5b6b18a2b6 : SUCCESS in 3m 04s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/d23154399d404949ad4dc366e7b19355 : FAILURE in 3m 14s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/f696821540c24207b2dfe48a3452c722 : FAILURE in 9m 38s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/6d5fd716cabb417d906344173a054700 : SUCCESS in 1m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1e2ca5ee7d344a03be5ab5fc80db6538 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":9},{"id":"85c5ac6ab758f9199d94da652885d5dd23005897","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 22:20:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"aa789659ce33c3afd2425a5aa5b508f376ff2d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:43:16.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5afe33cbf0a451b92c5730b2e67b4df\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/304501cb8e144db6a2220bdd09600b5c : SUCCESS in 2m 07s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/41fd20baa5e348db87fde1b7ca719093 : SUCCESS in 2m 53s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/45a8c6f185a549b187c6bf3c180521e0 : SUCCESS in 17m 16s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/eac3f421698c4d419be5f0192c1adc78 : FAILURE in 8m 27s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/ca9e99da80454f8d94fa3761ae70a4c9 : SUCCESS in 1m 33s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1621c1539c5f47fb9a2d8b924ee65070 : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":9},{"id":"0d00ed295c692b70341adcdf5cfc062562056907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 22:57:35.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0083d8207a239e3750fb9002a6b881e57163a7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:16:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4559b30668b4c3194c8af8937fc8493\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/7dacd68ddc5c45859dc36bcadf89cf5f : SUCCESS in 2m 09s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6d831502ce724a23b82bc642c40f1bc6 : SUCCESS in 3m 00s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/120d148814e2468c829ab0b64082453e : FAILURE in 2m 36s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/8cdee63d542243c291095a2fab7d45f5 : SUCCESS in 16m 50s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/189de5d227e94b15ad945c48bec02fa2 : SUCCESS in 1m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/00da6ff9e219445ea93f814470b7b98f : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":10},{"id":"860f0779b674ec3f9143779208303589edce298c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 23:21:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"35a93873494463a73e1aa670084b90891d52a050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:33:29.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2172963723c846f5a13bcc3f2cbbd1a0\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/4e8a2a046ca243cbaf8fc1dbcf64f876 : SUCCESS in 3m 01s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b362edf408aa4ab6a9152a5ce21c2f12 : SUCCESS in 4m 07s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/3042f45844254d69b7efa64e25bb3f8d : FAILURE in 2m 23s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/87ebe1126d354791b18f07dd85b09213 : FAILURE in 9m 18s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/27ce66a852af431e8daad53ee6edf14e : SUCCESS in 1m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d05b46bbe07948cdbc5a4ca798f7f454 : FAILURE in 3m 21s","accounts_in_message":[],"_revision_number":10},{"id":"2507f8ad2774bf0b29b053ee85144223e162af54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-21 23:55:43.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"afe3744b5e672f511bcb479c0414494676da3783","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 00:01:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3511ec1803680194f4e69d172148b442f2b21032","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 00:22:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d20ca3e395a476f960ee777936d919b\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/46c4f70be2c74fab90860e6d7b54a3db : SUCCESS in 2m 26s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6c0b0688fd0a4983a19b2fae18671ccf : SUCCESS in 3m 13s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/cdeac970e9324da6b93684689d73a7ed : SUCCESS in 17m 38s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/a72ac57f133c4dda85210ce6948c0d99 : SUCCESS in 19m 03s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/8da4ebccc3a6484f91261882242e7964 : SUCCESS in 1m 52s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eecb86cceb424070a1675ad356e71296 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":12},{"id":"f81fe5b62b1f8f4a84e3345b1ca334edba302196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 00:23:52.000000000","message":"Uploaded patch set 13: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"f37f634352fe18277c77735c61a7435c9a39aee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 00:34:59.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"f3c24eb5864b82ac7bd9f2d8950408d491f14ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 00:36:06.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d5bbff8c1ba0eaca8076f566321186b2aeea5333","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 01:46:44.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"499d5cb19162b6575db7ec6003e0293d08d6d8d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 01:51:11.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"cc66ff31206e5c88aae50c088f9da83db580af54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 02:38:06.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3096a8f1798541389386141a9f1d4ee8\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/23d98599c9674625851563c320312f14 : SUCCESS in 3m 08s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c0db53ea7aac431e9ab9d0c635ebdd0c : SUCCESS in 4m 22s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/3c5ee65412a5492993281bc88e0856c9 : SUCCESS in 17m 50s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/b985c3bdb22b4079a9097a57f9343edc : SUCCESS in 24m 26s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/53071b034f7f4f6fad44b118836f9d37 : SUCCESS in 2m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e74fb4d468644805bc72bd04732958f8 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":16},{"id":"5b80bd95e733703c11699686756a45deb751e87c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-11-22 16:20:03.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"7c86e6c38b8bf36cf120f1148fe5ae2ddad31e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 16:58:54.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbff5ac1c0c410fb84f645972f0ce30\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/f7683f54982e47058e55da8aa7af119e : SUCCESS in 2m 51s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a1738fda0aa4421aab69e0d067112db9 : SUCCESS in 4m 12s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/5de3c0a498404ea6b0c91744a8fb1960 : SUCCESS in 14m 35s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/47049a22e4c64afea7ce523fe7422d48 : SUCCESS in 21m 39s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/6688219754984c2383c491050af59b89 : SUCCESS in 2m 13s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0ad14bdd27004a7d89fdba9516a62774 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":16},{"id":"51ba3529dbe786d5698398d1bd3020e0d2b315c5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-11-22 19:32:39.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"afa0d9d8b08a97cd0465caa498b482d13c27f522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:33:08.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"640d49e8480cbff89edc3a461c158ee04af28088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:54:54.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38d347b558db42c3a1ba64a843cba9ff\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/e793625999e0440f845f3e9abb998c54 : SUCCESS in 1m 59s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/cada13a0a4414baca0d3c834d4ae0eb9 : SUCCESS in 16m 05s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/2ed204f069ee44c880136a93d493f9fb : SUCCESS in 17m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0622d8eaa68948509788c5972de5504c : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":16},{"id":"397edae8d622678787e33894da83608892d5ffba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:54:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"34ac6bcdc426d033368e97f4890d4ca1c6e0d576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 19:57:48.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c22b8d308ea8498aac718bf8b44b0bf8\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/ff84de5d39634a95b6d5d75bf5ce93b4 : SUCCESS in 37s\n- openafs-rpm-package-build-promote https://zuul.opendev.org/t/openstack/build/4fae7435320649c38bcbf8b8d0a17d52 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"e89e5dcb1515372afe4c85c0e60388a9cd091bb0","revisions":{"d4f7f9ab75228b3f269fd043f953776c11154ccd":{"kind":"REWORK","_number":1,"created":"2024-11-20 17:12:42.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/1"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d4f7f9ab75228b3f269fd043f953776c11154ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d4f7f9ab75228b3f269fd043f953776c11154ccd"}]},"branch":"refs/heads/master"},"37e48b6d709af3bcf87aef474bf427310e785b2b":{"kind":"REWORK","_number":2,"created":"2024-11-20 17:18:02.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/2"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:17:57.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/37e48b6d709af3bcf87aef474bf427310e785b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/37e48b6d709af3bcf87aef474bf427310e785b2b"}]},"branch":"refs/heads/master"},"f4117e44c12ca06131bf67387a76ef7a35a22b01":{"kind":"REWORK","_number":3,"created":"2024-11-21 00:54:33.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/3"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 00:54:31.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f4117e44c12ca06131bf67387a76ef7a35a22b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f4117e44c12ca06131bf67387a76ef7a35a22b01"}]},"branch":"refs/heads/master"},"a3f6d53c52ed2a22bec0c39bb1aaaff72fc2aef6":{"kind":"REWORK","_number":4,"created":"2024-11-21 18:04:15.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/4"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 18:04:03.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a3f6d53c52ed2a22bec0c39bb1aaaff72fc2aef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a3f6d53c52ed2a22bec0c39bb1aaaff72fc2aef6"}]},"branch":"refs/heads/master"},"27d08b60a6ee243ad04710ffff7debe33055cb4e":{"kind":"REWORK","_number":5,"created":"2024-11-21 18:13:56.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/5"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 18:13:50.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/27d08b60a6ee243ad04710ffff7debe33055cb4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/27d08b60a6ee243ad04710ffff7debe33055cb4e"}]},"branch":"refs/heads/master"},"dc068042dc41b0be574b6a5d5c1007a9db43d7ab":{"kind":"REWORK","_number":6,"created":"2024-11-21 19:02:09.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/6"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 19:01:42.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/dc068042dc41b0be574b6a5d5c1007a9db43d7ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/dc068042dc41b0be574b6a5d5c1007a9db43d7ab"}]},"branch":"refs/heads/master"},"0126c85d3156c7b2c32a79315571e475f3dcd656":{"kind":"REWORK","_number":7,"created":"2024-11-21 19:12:20.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/7"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 19:12:15.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0126c85d3156c7b2c32a79315571e475f3dcd656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/0126c85d3156c7b2c32a79315571e475f3dcd656"}]},"branch":"refs/heads/master"},"6f05b9127a7ebc34ab845b3cf3f36b980689087b":{"kind":"REWORK","_number":8,"created":"2024-11-21 19:19:56.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/8"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 19:19:50.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6f05b9127a7ebc34ab845b3cf3f36b980689087b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6f05b9127a7ebc34ab845b3cf3f36b980689087b"}]},"branch":"refs/heads/master"},"64fb2a466b280bae07f9cd48b822395bf050cba5":{"kind":"REWORK","_number":9,"created":"2024-11-21 21:59:49.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/9"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 21:59:42.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/64fb2a466b280bae07f9cd48b822395bf050cba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/64fb2a466b280bae07f9cd48b822395bf050cba5"}]},"branch":"refs/heads/master"},"c22803b30c25abfeedaf8aa3432ac1fc0ea39819":{"kind":"REWORK","_number":10,"created":"2024-11-21 22:57:35.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/10"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 22:57:30.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c22803b30c25abfeedaf8aa3432ac1fc0ea39819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c22803b30c25abfeedaf8aa3432ac1fc0ea39819"}]},"branch":"refs/heads/master"},"cf52e23a9f6db42ce08ef98371587b4f43dcacc3":{"kind":"REWORK","_number":11,"created":"2024-11-21 23:55:43.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/11"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-21 23:54:31.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nWe also manually select the new kernel with grubby as the default kernel\nto boot as we have had problems with arm64 efi nodes sometimes booting\non the old kernel. Also switch package module installs to dnf module\ninstalls since all of our installations are occuring on CentOS 9 or\nnewer.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/cf52e23a9f6db42ce08ef98371587b4f43dcacc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/cf52e23a9f6db42ce08ef98371587b4f43dcacc3"}]},"branch":"refs/heads/master"},"580337bd942f2d2b8bab5449830a2da5cc28a900":{"kind":"REWORK","_number":12,"created":"2024-11-22 00:01:21.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/12"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-22 00:01:16.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nWe also manually select the new kernel with grubby as the default kernel\nto boot as we have had problems with arm64 efi nodes sometimes booting\non the old kernel. Also switch package module installs to dnf module\ninstalls since all of our installations are occuring on CentOS 9 or\nnewer.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/580337bd942f2d2b8bab5449830a2da5cc28a900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/580337bd942f2d2b8bab5449830a2da5cc28a900"}]},"branch":"refs/heads/master"},"525a578a8d360925ef48718c382b365bbfb10e92":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-11-22 00:23:52.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/13"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-22 00:20:36.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nWe also manually select the new kernel with grubby as the default kernel\nto boot as we have had problems with arm64 efi nodes sometimes booting\non the old kernel. This is likely due to this issue:\n\n  https://bugzilla.redhat.com/show_bug.cgi?id\u003d2032680\n\nAlso switch package module installs to dnf module installs since all of\nour installations are occuring on CentOS 9 or newer.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/525a578a8d360925ef48718c382b365bbfb10e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/525a578a8d360925ef48718c382b365bbfb10e92"}]},"branch":"refs/heads/master"},"aedf5f8cc5e9deaadb43c0c93a51d1a2aae8fd53":{"kind":"REWORK","_number":14,"created":"2024-11-22 00:34:59.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/14"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-22 00:34:08.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nWe also manually select the new kernel as the default kernel and remove\nthe old kernel with grubby as dnf doesn\u0027t reliably update the kernel to\nboot automatically on arm64 centos 9 images. This is likely due to this\nissue:\n\n  https://bugzilla.redhat.com/show_bug.cgi?id\u003d2032680\n\nAlso switch package module installs to dnf module installs since all of\nour installations are occuring on CentOS 9 or newer.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/aedf5f8cc5e9deaadb43c0c93a51d1a2aae8fd53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/aedf5f8cc5e9deaadb43c0c93a51d1a2aae8fd53"}]},"branch":"refs/heads/master"},"3e8d72e7cbd6a70771137a480820a2956c1fb40f":{"kind":"REWORK","_number":15,"created":"2024-11-22 00:36:06.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/15"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-22 00:36:00.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nWe also manually select the new kernel as the default kernel and remove\nthe old kernel with grubby as dnf doesn\u0027t reliably update the kernel to\nboot automatically on arm64 centos 9 images. This is likely due to this\nissue:\n\n  https://bugzilla.redhat.com/show_bug.cgi?id\u003d2032680\n\nAlso switch package module installs to dnf module installs since all of\nour installations are occuring on CentOS 9 or newer.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3e8d72e7cbd6a70771137a480820a2956c1fb40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/3e8d72e7cbd6a70771137a480820a2956c1fb40f"}]},"branch":"refs/heads/master"},"e89e5dcb1515372afe4c85c0e60388a9cd091bb0":{"kind":"REWORK","_number":16,"created":"2024-11-22 01:51:11.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/13/935813/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/13/935813/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/13/935813/16"}}},"commit":{"parents":[{"commit":"b13850cd49c3cddefd1c1ea965b5792e9c61ac20","subject":"Merge \"Update CentOS 9 Stream OpenAFS package to 1.8.13\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b13850cd49c3cddefd1c1ea965b5792e9c61ac20"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-20 17:11:38.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-11-22 01:51:03.000000000","tz":-480},"subject":"Update system packages and reboot when building centos openafs","message":"Update system packages and reboot when building centos openafs\n\nWe often hit issues with our CentOS arm images being slightly stale.\nWhen this happens we get openafs package build errors because available\nheaders don\u0027t match the running kernel. Address this by updating the\nentire running system and rebooting onto any new kernel before we build\nthe openafs package. This should ensure our local install matches\nupstream packages when it comes to kernel packages.\n\nWe also manually select the new kernel as the default kernel and remove\nthe old kernel with grubby as dnf doesn\u0027t reliably update the kernel to\nboot automatically on arm64 centos 9 images. This is likely due to this\nissue:\n\n  https://bugzilla.redhat.com/show_bug.cgi?id\u003d2032680\n\nAlso switch package module installs to dnf module installs since all of\nour installations are occuring on CentOS 9 or newer.\n\nChange-Id: Idb42386071d6652312bb343923473a62c55e2828\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/e89e5dcb1515372afe4c85c0e60388a9cd091bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/e89e5dcb1515372afe4c85c0e60388a9cd091bb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
