)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~950768","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Ia12593821f95679e794457a56b58b4ca004d367a","project":"openstack/openstack-zuul-jobs","branch":"master","attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-07-08 19:36:08.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":["generated-by"],"change_id":"Ia12593821f95679e794457a56b58b4ca004d367a","subject":"Comply with Policy for AI Generated Content","status":"MERGED","created":"2025-05-23 15:49:01.000000000","updated":"2025-07-08 19:46:57.000000000","submitted":"2025-07-03 13:13:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950768","meta_rev_id":"2b8d732296b984f08653ef3253fc8edea3d824c6","_number":950768,"virtual_id_number":950768,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-03 13:13:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2025-05-23 15:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-07-03 13:00:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-23 15:50:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2025-05-23 15:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-08 19:36:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"6aa2c47179fa7049b70e2aa765e93193cba6fa42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-23 15:49:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5635b0a3090f68dee1e33016ed17798ab270cfa7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-05-23 15:50:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"261a01e1ae5533badf90c2df0dd9fd6a0d4d97f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 15:58:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd47e896d45142c9920523360eb377c7\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/75c38b610b80499b859564e0c36ebd52 : SUCCESS in 2m 22s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c87d68dced02440484937e8e26d134b0 : SUCCESS in 3m 05s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/618282f0686b41cfaf60c289ca368705 : SUCCESS in 1m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eddb5a47ed924a09b917e2cc20116173 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"793dec14f79570ce6a6b71f1592324c218c7f3ba","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-23 16:13:07.000000000","message":"Hashtag added: generated-by","accounts_in_message":[],"_revision_number":1},{"id":"f6b2b9c71ed6b0597ed0ee39931bb7b026aac2d1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-30 19:51:48.000000000","message":"Patch Set 1: Workflow+1\n\nSelf-approving to keep these consistent with changes generated by scripts in openstack/releases.","accounts_in_message":[],"_revision_number":1},{"id":"132972e87651d5c779c891f4dcc65d6fba0deb16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 19:52:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"82ce18d12c9df313a3c02bd51cae9bb1f7c7d255","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-30 19:53:09.000000000","message":"Patch Set 1: -Workflow\n\nNever mind, that change hasn\u0027t merged yet.","accounts_in_message":[],"_revision_number":1},{"id":"5011b73c04a8f78b50b26fc1cc4b9e3bcc391fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 19:53:11.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"db0354f842e83350b1b8dddce239b58718e05f7e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-06-30 19:58:45.000000000","message":"Patch Set 1:\n\nrecheck for fresh test results","accounts_in_message":[],"_revision_number":1},{"id":"97644a08f778d2c8d002528f28564da1c5926706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 20:07:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042c3a53be1d4172a94d50cb08e77fca\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/5ae445bb00d64fda8f572cb801ed766f : SUCCESS in 2m 28s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/71cb1c022f0940c7a5e05e04fca37a19 : SUCCESS in 2m 59s\n- openstack-infra-extra-integration-bionic https://zuul.opendev.org/t/openstack/build/e3e56c4517034683bf09501f6f3745ac : SUCCESS in 1m 49s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d4baa541b1d43aa96f92e9d59f95a65 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a48e0b1e305d04233b7dd73215ff51cd0c97a2d9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-07-03 13:00:15.000000000","message":"Patch Set 1: Workflow+1\n\nThe similar openstack/releases change Id798bba638543cf8e9573c10e85495e7cc436237 merged, so let\u0027s go ahead and land this too for consistency.","accounts_in_message":[],"_revision_number":1},{"id":"2c0344c1ed985251a29025d3469ffae8a6ef8d52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:02:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1af09eeeae48627328f778018873302ee33e0fb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:13:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b26ea3a9a6476e994a59f9c4b3989f\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/d9941b98d0094c2a9207a73f9583fa05 : SUCCESS in 2m 38s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/db0927329e33496896127078b5931e5e : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d73929bffa250c1cb47ae0017ca5f8adac48cf2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:13:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"60904fb80ac1309d6ea2b490be45f24210053fc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 13:14:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8da00a410eaf478eae80935a21843d0c\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/7c1b4383c053442d814a28418fdd5cc3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"66102918bc11672634b89357a70a98f42292e03d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 19:36:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b8d732296b984f08653ef3253fc8edea3d824c6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-08 19:46:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"70fa0bd649fe39ee36130bbfeeefc08325fcfc6c","revisions":{"70fa0bd649fe39ee36130bbfeeefc08325fcfc6c":{"kind":"REWORK","_number":1,"created":"2025-05-23 15:49:01.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/68/950768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/68/950768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/68/950768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/68/950768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/68/950768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/68/950768/1"}}},"commit":{"parents":[{"commit":"422c1d43d49a4633a07c8fcecb66e3e13094c4cb","subject":"Make scripts DCO compliant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/422c1d43d49a4633a07c8fcecb66e3e13094c4cb"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-05-23 15:43:38.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-05-23 15:43:38.000000000","tz":0},"subject":"Comply with Policy for AI Generated Content","message":"Comply with Policy for AI Generated Content\n\nWhile probably not meeting most people\u0027s definition of artificial\nintelligence these scripts do create patches. The OpenInfra Policy\nfor AI Generated Content recommends to:\n\n\u003e Add a \"Generated-By:\" label to the commit message, and explain in\n\u003e comments or the commit message any prompts or background context\n\u003e the reviewers might need to fully understand the change and how\n\u003e much of the change was generated by the tool.\n\nEven though it also has a carve-out for these sorts of scripts:\n\n\u003e The exception to the rule is that we do allow submissions from\n\u003e well documented automated processes, such as release tooling or\n\u003e for internationalization updates.\n\nhttps://openinfra.org/legal/ai-policy\n\nNevertheless, documenting within the commit message the location of\nthe script which was used to generate that contribution is helpful\nfor a number of obvious reasons. Let\u0027s do it anyway, even though\nit\u0027s not technically required.\n\nChange-Id: Ia12593821f95679e794457a56b58b4ca004d367a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/70fa0bd649fe39ee36130bbfeeefc08325fcfc6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/70fa0bd649fe39ee36130bbfeeefc08325fcfc6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
