)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~951201","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I7448b2d9d24925f85b6c31eff53b323a7170f2df","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"openafs-18.13.2-rpm","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-05-28 22:13:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7448b2d9d24925f85b6c31eff53b323a7170f2df","subject":"Build openafs 1.8.13.2 rpm packages","status":"MERGED","created":"2025-05-28 19:58:21.000000000","updated":"2025-05-28 22:14:13.000000000","submitted":"2025-05-28 22:13:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951201-openafs-18.13.2-rpm","meta_rev_id":"99779e3489df5e866e81d5ae3a773a4c6f33ea7b","_number":951201,"virtual_id_number":951201,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-28 22:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-28 19:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-28 19:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 19:58:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-05-28 20:46:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d16703aa28a6e247c971a3bfd10e8ee7bd19773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-05-28 19:58:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e418b6b77f0a7fec79ea602ac1e8323917a7421","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-05-28 19:58:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3048432a93dfc5362891ca872ef19fe1a5f213f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 20:46:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27844405aa5b4172b4214af2b5dca2d5\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/68414830acb14ffd9d148b7b34d961a3 : SUCCESS in 2m 05s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8be802da93084562bb76aa8523228617 : SUCCESS in 7m 45s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/9d23572b45d74586bde3e95959bb0834 : SUCCESS in 18m 17s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/605e4ae9191349ccbd80400c75a647dd : SUCCESS in 13m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/002197dbe0fa4b07b5cdcdec3832267c : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"75d50c5e15ddf84b1403b3e938f474a87f876585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 20:46:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac951d1cc394588bd61c1272e5e2a6560e1f73e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 22:13:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c93981ab357f4a75a28a13c3b22eecfa\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/c312de8a9f7240c7b36a2ef00401c300 : SUCCESS in 3m 44s\n- openafs-rpm-package-build-centos-9-stream-x86 https://zuul.opendev.org/t/openstack/build/58fa9c03b2de4b33acb2699558373853 : SUCCESS in 17m 59s\n- openafs-rpm-package-build-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/ff5e35093bca4f8492590585639cce83 : SUCCESS in 23m 23s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c292e97bd9ad4f5881bde6e5d52bcd10 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"861d5ddca64428e4cc0b3b6f6365b6f4f845bdc6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 22:13:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"99779e3489df5e866e81d5ae3a773a4c6f33ea7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 22:14:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2224418a05b420db179d05344caa832\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/4f36a455266642d2b90dd4a3f9880231 : SUCCESS in 33s\n- openafs-rpm-package-build-promote https://zuul.opendev.org/t/openstack/build/8c2f32a1a5c94e75a7fbf8f4a9971bfb : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1cd58446b1f1ba6fa91ac2b91d28f930bae3182","revisions":{"d1cd58446b1f1ba6fa91ac2b91d28f930bae3182":{"kind":"REWORK","_number":1,"created":"2025-05-28 19:58:21.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/01/951201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/01/951201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/951201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/951201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/951201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/01/951201/1"}}},"commit":{"parents":[{"commit":"f1140fe7bab6ac124ebd4d68dee832d26462faec","subject":"Remove 2023.2 from translation job template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/f1140fe7bab6ac124ebd4d68dee832d26462faec"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-05-28 19:57:04.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-05-28 19:57:04.000000000","tz":-420},"subject":"Build openafs 1.8.13.2 rpm packages","message":"Build openafs 1.8.13.2 rpm packages\n\nLooks like the current version of our 1.8.13 packages don\u0027t get along\nwith DKMS on CentOS stream 9 due to the use of NO_WEAK_MODULES [0].\nAttempt to build 1.8.13.2 to see if the newer version addresses this.\n\n[0] https://paste.openstack.org/show/bnfVKVLwxCzq9bsLzqYb/\n\nChange-Id: I7448b2d9d24925f85b6c31eff53b323a7170f2df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d1cd58446b1f1ba6fa91ac2b91d28f930bae3182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d1cd58446b1f1ba6fa91ac2b91d28f930bae3182"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
