)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~968398","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~Idd60a479ff8bdd5a2e3161b1a8e7245ede3eb607","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"tox-releasenotes","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-06-17 03:49:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-06-16 18:00:46.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-16 17:52:36.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2026-05-12 18:59:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13177\u003e replied on the change","reason_account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-05-13 15:36:03.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Idd60a479ff8bdd5a2e3161b1a8e7245ede3eb607","subject":"Re-add translation support to build-openstack-releasenotes","status":"MERGED","created":"2025-11-25 20:58:45.000000000","updated":"2026-06-17 03:49:36.000000000","submitted":"2026-06-16 18:00:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"968398-tox-releasenotes","meta_rev_id":"b559c668ffa783c00071e202b3d05d7c5fb54279","_number":968398,"virtual_id_number":968398,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 18:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2026-05-09 13:23:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2026-06-16 17:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-12 18:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-12 21:06:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-05-13 15:36:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2026-06-16 17:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-25 21:04:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-27 18:20:27.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2026-05-09 13:23:12.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2026-05-12 21:06:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-05-12 21:06:31.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-05-16 21:47:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-06-16 17:24:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-17 03:49:36.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"e9cf7affc0d0ba2f55b27922fdd9aee798c94dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 20:58:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b706aafa17d411b8ba062b201027dec5b698b390","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 21:04:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b104c63c00284681921d9427438986c5\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0a844a64d3f143afb40096781a597ab9 : SUCCESS in 1m 47s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/481cb70915704dc2a777eb26e4ccffbb : SUCCESS in 3m 05s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/2ad2dcbfef6d44c49de93f5e093b7a10 : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a4c8f794b964ee5ab37776e3a1dce93 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e39699a3e4731c4265356606dd8127307179b800","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 12:19:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f37387113a9b20fbbac5ff1c7a062838b73f96a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 12:35:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3005384221625b9f57a9f220a94453015c641dd4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 12:37:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3ffb325567d323b49abb6eea87521b0f6560d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 12:48:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7197de8589c546598964f55a1627fe39\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/b6778c4d62824d0b81a7c5785a8ea46b : SUCCESS in 2m 16s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/4e8c11cce2d5455ba5cd6cfea09a4b3f : SUCCESS in 2m 08s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/3c84ef7070e94ecebef8387ccfecd410 : SUCCESS in 1m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/43b05a8c1b9742dc8610010b4076bdc2 : FAILURE in 3m 07s","accounts_in_message":[],"_revision_number":2},{"id":"2a55e8279a2a0a410ac620fa9a7df715e05ca443","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2025-11-27 18:20:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c72e89ead54097c10571dc1d593234d355978d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 18:27:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5c5ae46cbde86309f8b0fc8cf9dacf74c2315458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-27 18:37:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84204b7dc6fb434f921126d74c161f96\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fcb22bd40a7648f0a9cb3965e822b094 : SUCCESS in 1m 39s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0b5d18cb54d0496fa5771709da95f48a : SUCCESS in 3m 12s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/6ede362db5f1421aa9d4dd2f35416e06 : SUCCESS in 1m 17s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/56b50bc3bbea45f3a4156f041ba92c68 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"9e6ea9150d64b23344d3d1a415cc03e64d34882f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 18:43:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"515e7fb7ce290ea2d8df08285221aa9b8b6f168d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-27 18:50:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9a72de8e26c990e981905f3e0e9c81d1fe95be58","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-01-13 12:43:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6fd4d486fdd50a12d053f522d9d102c2313a4e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-08 18:04:15.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"651b92bb63d6ffa3fb497bd9fc64a226b0a40447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-08 18:19:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae74c6238f24784b1475940bf5cb8cb\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/369ecede05fa43418a7b93b9907b73ce : SUCCESS in 2m 36s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a06cc0f95a044262809264480f00f820 : SUCCESS in 3m 57s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/0a65b8f9e9e24891b05c7228fef88d89 : SUCCESS in 3m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/36202f4d72254399aab0193a988a729a : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2e23abdd1da71311af1a7c53cf69b25f026351d5","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2026-05-09 13:23:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a26bc5e71a6899cea21261f391d81d1d1f1e11b7","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2026-05-12 18:59:54.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"fe22d5f84c4e32160a06a4e116b450f45abc4cc0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-12 21:06:31.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1b664a2ddaf74c44f6c71059a6a0258a5a31fc22","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-13 15:36:03.000000000","message":"Patch Set 4: Code-Review+2\n\nI guess Ansible knows to install scripts from the role\u0027s files/ directory into the execution search path without a separate task for doing so? Typically I would expect to see this sort of solution implemented as a custom Ansible module instead, but I don\u0027t really have an opinion on which is better.","accounts_in_message":[],"_revision_number":4},{"id":"0bfa46b5f425bef6a800f2c10cb6af217757e3f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-16 17:52:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d715710728211745f1698dffb1e04e4ce1fba7e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:53:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d99f4d33fc269e2f8758163d7d31c64f392974f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:00:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aefbe359e9324e1d840e7b79c87cc3d5\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/fd5884ee8a1a46aebc095b2e4bd5fd98 : SUCCESS in 2m 38s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8cc91912bd6d40d3b7f61a8b439fa0ab : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":4},{"id":"533f216bb97bf6b75a79dbfd7a321504090c7814","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:00:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"81a72a3945e0d8e251d0ae8f8153f67cbe6a7d29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 18:03:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d244af44bb4c48ac6983f527d1238b\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/c7900fcfeb124890abc92c9e1e3c5879 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":4},{"id":"b93690c855ff508ea167cb787b9313e3de1f9e29","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-16 22:40:07.000000000","message":"Patch Set 4:\n\nNote that this and a subsequent attempt to fix it were reverted in change Ic6aafb92d457115a5c6f90eba078fac66cb75ca1 in order to get release notes jobs passing again for now.","accounts_in_message":[],"_revision_number":4},{"id":"b559c668ffa783c00071e202b3d05d7c5fb54279","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-17 03:49:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d2e10dfe578f0c7c3468a35435129d85635dfa6b","revisions":{"22cb9581a64dfb1be9f60b511bb2550a084ac4d9":{"kind":"REWORK","_number":1,"created":"2025-11-25 20:58:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/968398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/98/968398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/1"}}},"commit":{"parents":[{"commit":"b07bf5b4b0d1df18672c9a5c09709d910b2e841e","subject":"Switch to building releasenotes with tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b07bf5b4b0d1df18672c9a5c09709d910b2e841e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-25 20:41:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-25 20:57:12.000000000","tz":0},"subject":"Re-add translation support to build-openstack-releasenotes","message":"Re-add translation support to build-openstack-releasenotes\n\nMigrate the logic that currently exist in zuul/zuul-jobs in here, with\nan extra bit of glue to ensure we use the correct \u0027env_dir\u0027 (since some\njobs might still be overriding this to use \u0027docs\u0027 even though that no\nlonger does the right thing in tox 4.x).\n\nChange-Id: Idd60a479ff8bdd5a2e3161b1a8e7245ede3eb607\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/22cb9581a64dfb1be9f60b511bb2550a084ac4d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/22cb9581a64dfb1be9f60b511bb2550a084ac4d9"}]},"branch":"refs/heads/master"},"2e0df4be8212ba054f97f58874288e58d32994b1":{"kind":"REWORK","_number":2,"created":"2025-11-27 12:35:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/968398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/98/968398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/2"}}},"commit":{"parents":[{"commit":"b07bf5b4b0d1df18672c9a5c09709d910b2e841e","subject":"Switch to building releasenotes with tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b07bf5b4b0d1df18672c9a5c09709d910b2e841e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-25 20:41:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-27 12:34:57.000000000","tz":0},"subject":"Re-add translation support to build-openstack-releasenotes","message":"Re-add translation support to build-openstack-releasenotes\n\nMigrate the logic that currently exist in zuul/zuul-jobs in here, with\nan extra bit of glue to ensure we use the correct \u0027env_dir\u0027 (since some\njobs might still be overriding this to use \u0027docs\u0027 even though that no\nlonger does the right thing in tox 4.x).\n\nChange-Id: Idd60a479ff8bdd5a2e3161b1a8e7245ede3eb607\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2e0df4be8212ba054f97f58874288e58d32994b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2e0df4be8212ba054f97f58874288e58d32994b1"}]},"branch":"refs/heads/master"},"48434c3a4207f035af13516ddfbfb2a6e3b173a6":{"kind":"REWORK","_number":3,"created":"2025-11-27 18:27:59.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/968398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/98/968398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/3"}}},"commit":{"parents":[{"commit":"b07bf5b4b0d1df18672c9a5c09709d910b2e841e","subject":"Switch to building releasenotes with tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b07bf5b4b0d1df18672c9a5c09709d910b2e841e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-25 20:41:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-27 18:27:42.000000000","tz":0},"subject":"Re-add translation support to build-openstack-releasenotes","message":"Re-add translation support to build-openstack-releasenotes\n\nMigrate the logic that currently exist in zuul/zuul-jobs in here, with\nan extra bit of glue to ensure we use the correct \u0027env_dir\u0027 (since some\njobs might still be overriding this to use \u0027docs\u0027 even though that no\nlonger does the right thing in tox 4.x).\n\nChange-Id: Idd60a479ff8bdd5a2e3161b1a8e7245ede3eb607\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/48434c3a4207f035af13516ddfbfb2a6e3b173a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/48434c3a4207f035af13516ddfbfb2a6e3b173a6"}]},"branch":"refs/heads/master"},"d2e10dfe578f0c7c3468a35435129d85635dfa6b":{"kind":"REWORK","_number":4,"created":"2026-05-08 18:04:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/968398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/98/968398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/98/968398/4"}}},"commit":{"parents":[{"commit":"b07bf5b4b0d1df18672c9a5c09709d910b2e841e","subject":"Switch to building releasenotes with tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/b07bf5b4b0d1df18672c9a5c09709d910b2e841e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-11-25 20:41:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-05-08 18:04:14.000000000","tz":0},"subject":"Re-add translation support to build-openstack-releasenotes","message":"Re-add translation support to build-openstack-releasenotes\n\nMigrate the logic that currently exist in zuul/zuul-jobs in here, with\nan extra bit of glue to ensure we use the correct \u0027env_dir\u0027 (since some\njobs might still be overriding this to use \u0027docs\u0027 even though that no\nlonger does the right thing in tox 4.x).\n\nChange-Id: Idd60a479ff8bdd5a2e3161b1a8e7245ede3eb607\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d2e10dfe578f0c7c3468a35435129d85635dfa6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/d2e10dfe578f0c7c3468a35435129d85635dfa6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
