)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~969949","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I34874b46ef70b573eab151cfdb47125c0b133d08","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"caracal-unmaintained","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-01-29 11:46:09.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-01-29 11:46:09.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-01-27 14:19:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I34874b46ef70b573eab151cfdb47125c0b133d08","subject":"Update 2024.1 Caracal related filters","status":"MERGED","created":"2025-12-05 15:13:02.000000000","updated":"2026-01-29 11:47:53.000000000","submitted":"2026-01-29 11:46:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"969949-caracal-unmaintained","meta_rev_id":"d8415b4221083702d43d22d374174e80ac3ef5c7","_number":969949,"virtual_id_number":969949,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 11:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-01-29 11:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2026-01-27 14:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2026-01-29 11:37:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-05 15:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-05 15:13:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-05 16:29:48.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-01-26 19:14:32.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-01-29 11:37:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5f80610e74cdfda0496175e2939fcffe65cda585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-05 15:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2883483d0cbfa73ea918fb1e61b438a84d95143b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:13:35.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/openstack-zuul-jobs on branch master.  The\nproblem was:\n\n  The key \"required-projects\" appears more than once; duplicate keys are not permitted.\n\n  branches:\n                - stable/2024.2\n                - stable/2025.1\n                - stable/2025.2\n              required-projects:\n                - openstack/neutron\n              required-projects:\n                - openstack/neutron\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master\", line 2730","accounts_in_message":[],"_revision_number":1},{"id":"06b5b5fcf1050b307a3ebf14ea66577e5e3adfbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:13:31.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/openstack-zuul-jobs on branch master.  The\nproblem was:\n\n  The key \"required-projects\" appears more than once; duplicate keys are not permitted.\n\n  branches:\n                - stable/2024.2\n                - stable/2025.1\n                - stable/2025.2\n              required-projects:\n                - openstack/neutron\n              required-projects:\n                - openstack/neutron\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master\", line 2730","accounts_in_message":[],"_revision_number":1},{"id":"b868803b259a185e5f307d4491afd1b0f7930adc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-05 15:18:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4c1b67389912e1611be379a78d4428c7f6e8bfcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:26:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b819d6757db47bca1d0c9d66d55221f\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/234ec64df1fd4842bf7c20c4dd61f1dc : SUCCESS in 3m 46s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/85cb50ed0aca48518361fdcb8f369ce2 : SUCCESS in 2m 56s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/e671765c6ec74d57a8710ad99ec01cde : SUCCESS in 1m 22s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f93829eef82246c78a7f89630621b5a0 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a6cff046b2424e146cca46d98e1c410dce172f00","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-12-05 16:29:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6e106522e688be58c2ad37557f43551d6707ec9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-01-26 19:14:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f0a64fdac4ca04eceef8c6fedb3b5b8aefb5e935","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-01-27 08:08:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c06f23881c3e2f93a2bfd040426933a99fc2a616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:16:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd4e54caf25f4ecdbcdee308295fca41\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/0349d2b7e569490f9fa2afef2235141d : SUCCESS in 2m 32s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e2136ae78c1b4c20afd311a6a8da3a3b : SUCCESS in 4m 03s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/16e7e77dca4f4a77b490181366e85c7a : SUCCESS in 1m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/12556b4544bf47d9877d2176d4c392e2 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"f97288d05629b3c0aa0ab58a41fa3ac1fdf8daff","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-01-27 14:19:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"df2d953fc3d347a310f2fadc6b0c8d93c49a92d2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-01-29 11:37:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e187452c095ea4ec8ca3078fe91adbde62ef0106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:37:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9c4b8a8f3c6176c8ccbe774a70a05ca65b4fbb9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:46:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c5f63478894c259d25be0a3ba1a770\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/01c9cef31b2a4282b59e1acfedcf2d02 : SUCCESS in 2m 40s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/927a441bf3cd404f8a315ab744e2c2df : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":3},{"id":"5ae4f1435b8b9379673ce567351b6bc11d9b6f22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:46:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d8415b4221083702d43d22d374174e80ac3ef5c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:47:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/184f64341f684d1eb6cfe3267ccd58a2\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/1da30da2109c44a686d8967c39324444 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc14eff6cac3a8beddd5e82bb825cf9cd4e46187","revisions":{"17d77c5780d5c7fbc97d661e2594936f8e2c44c5":{"kind":"REWORK","_number":1,"created":"2025-12-05 15:13:02.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/969949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/969949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/1"}}},"commit":{"parents":[{"commit":"a0826e168eae5bb4cf283a8d69e09a4731a6d7cf","subject":"Remove arm64 unit tests job from stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a0826e168eae5bb4cf283a8d69e09a4731a6d7cf"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-12-05 15:08:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-12-05 15:12:57.000000000","tz":60},"subject":"Update 2024.1 Caracal related filters","message":"Update 2024.1 Caracal related filters\n\n2024.1 Caracal moved to Unmaintained state, hence, in job templates\nthe unmaintained/2024.1 branch filter needs to be used instead of\nstable/2024.1 from now on.\nFurthermore, periodic-stable job templates do not need to run against\n2024.1 Caracal branches anymore, so those filters will be removed.\n\nChange-Id: I34874b46ef70b573eab151cfdb47125c0b133d08\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/17d77c5780d5c7fbc97d661e2594936f8e2c44c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/17d77c5780d5c7fbc97d661e2594936f8e2c44c5"}]},"branch":"refs/heads/master"},"c819ed88e9aac13ab0e2156592565a3fb6ca4588":{"kind":"REWORK","_number":2,"created":"2025-12-05 15:18:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/969949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/969949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/2"}}},"commit":{"parents":[{"commit":"a0826e168eae5bb4cf283a8d69e09a4731a6d7cf","subject":"Remove arm64 unit tests job from stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/a0826e168eae5bb4cf283a8d69e09a4731a6d7cf"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-12-05 15:08:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-12-05 15:17:48.000000000","tz":60},"subject":"Update 2024.1 Caracal related filters","message":"Update 2024.1 Caracal related filters\n\n2024.1 Caracal moved to Unmaintained state, hence, in job templates\nthe unmaintained/2024.1 branch filter needs to be used instead of\nstable/2024.1 from now on.\nFurthermore, periodic-stable job templates do not need to run against\n2024.1 Caracal branches anymore, so those filters will be removed.\n\nChange-Id: I34874b46ef70b573eab151cfdb47125c0b133d08\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c819ed88e9aac13ab0e2156592565a3fb6ca4588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/c819ed88e9aac13ab0e2156592565a3fb6ca4588"}]},"branch":"refs/heads/master"},"dc14eff6cac3a8beddd5e82bb825cf9cd4e46187":{"kind":"REWORK","_number":3,"created":"2026-01-27 08:08:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/969949/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/969949/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/969949/3"}}},"commit":{"parents":[{"commit":"94f3e3da6990fb5084dd828221610e552f7f526b","subject":"Merge \"Remove unused ceilometer job template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/94f3e3da6990fb5084dd828221610e552f7f526b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-12-05 15:08:31.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-01-27 08:07:09.000000000","tz":60},"subject":"Update 2024.1 Caracal related filters","message":"Update 2024.1 Caracal related filters\n\n2024.1 Caracal moved to Unmaintained state, hence, in job templates\nthe unmaintained/2024.1 branch filter needs to be used instead of\nstable/2024.1 from now on.\nFurthermore, periodic-stable job templates do not need to run against\n2024.1 Caracal branches anymore, so those filters will be removed.\n\nChange-Id: I34874b46ef70b573eab151cfdb47125c0b133d08\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/dc14eff6cac3a8beddd5e82bb825cf9cd4e46187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/dc14eff6cac3a8beddd5e82bb825cf9cd4e46187"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
