)]}'
{"id":"openstack%2Fopenstack-zuul-jobs~984749","triplet_id":"openstack%2Fopenstack-zuul-jobs~master~I503dbda1bbea1e2fecbafdd02b2e36bd254cc2f7","project":"openstack/openstack-zuul-jobs","branch":"master","topic":"unmaintained-job-template-fix","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-06-16 17:46:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I503dbda1bbea1e2fecbafdd02b2e36bd254cc2f7","subject":"Add py310 job to openstack-python3-jobs for 2023.1","status":"MERGED","created":"2026-04-15 13:24:54.000000000","updated":"2026-06-16 17:50:02.000000000","submitted":"2026-06-16 17:46:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"984749-unmaintained-job-template-fix","meta_rev_id":"e4f93f98319c6f066ac7c32d4b11faa56b080fd7","_number":984749,"virtual_id_number":984749,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-16 17:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-05-03 13:39:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-06-16 17:14:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-06-16 17:14:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-15 13:25:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-15 18:30:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2026-05-03 13:39:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-16 17:14:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cd3ceda1efe9c9e2918d9e2299459ed8e2716318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-15 13:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f1e7fc181e4f578bdd095fececc3f849fc0f6b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 13:25:15.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/openstack-zuul-jobs on branch master.  The\nproblem was:\n\n  Job build-wheel-cache-ubuntu-bionic not defined\n\nThe problem appears in the \"opendev.org/openstack/openstack-zuul-jobs\" project stanza:\n\n  project:\n      templates:\n        - publish-tox-docs-infra\n  \n      check:\n        jobs:\n          - opendev-buildset-registry\n          # TODO(frickler): drop this again when latest tox is unbroken\n          - opendev-tox-docs:\n  ...\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"63d082849858b1c01304912c392539c83d78ed86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 13:25:32.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/openstack-zuul-jobs on branch master.  The\nproblem was:\n\n  Job build-wheel-cache-ubuntu-bionic not defined\n\nThe problem appears in the \"opendev.org/openstack/openstack-zuul-jobs\" project stanza:\n\n  project:\n      templates:\n        - publish-tox-docs-infra\n  \n      check:\n        jobs:\n          - opendev-buildset-registry\n          # TODO(frickler): drop this again when latest tox is unbroken\n          - opendev-tox-docs:\n  ...\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/project.yaml@master\", line 2","accounts_in_message":[],"_revision_number":1},{"id":"c064ec45d3710af59bab1d94176ec132790d81e7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-15 18:30:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3f6e6f0d81392082c4ac83c65efe6ebdc12e444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 18:39:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d2b921d2231414a96fdb25efff1231e\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/235e1737cfc6493d98c16551b6dbb3a7 : SUCCESS in 1m 54s\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/43994ee7ea54427e80f6f2470160ac78 : SUCCESS in 2m 37s\n- openstack-infra-extra-integration-noble https://zuul.opendev.org/t/openstack/build/53d0cd9583694c3393f5e3bc86a5a641 : SUCCESS in 2m 15s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fa7a3cd5b9f64b6994acc012cb736efb : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ad6da611f5e306418268178e03f35e07622a78a6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-05-03 13:39:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4eb7cdefece4422e124c3cea50076a2f4cd4e117","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-16 17:14:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"952db5ca03493144b0f547f858066e0837210e59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:20:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb47eda91ab852e21a54a5deca8a6dff285cccd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:46:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/614b09c1d11f423e8eabd9c194e4ae6a\n\n- opendev-tox-docs https://zuul.opendev.org/t/openstack/build/bbd9944bb8c54b19a0525cea9002c959 : SUCCESS in 2m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f86dfc9ce04446aba3f5a28ab94ff309 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5cd4a397b8899496cbbaa6af290ec3c42b2902bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:46:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4f93f98319c6f066ac7c32d4b11faa56b080fd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-16 17:50:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fc5d09c6371412c851a756f95d4208c\n\n- promote-tox-docs-infra https://zuul.opendev.org/t/openstack/build/b44f1a747da44c47accfa3d380e3d529 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ee11fc0f01c3e3828886645509bbbd36fe613b2","revisions":{"6ee11fc0f01c3e3828886645509bbbd36fe613b2":{"kind":"REWORK","_number":1,"created":"2026-04-15 13:24:54.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/984749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-zuul-jobs","ref":"refs/changes/49/984749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/984749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/984749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/984749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-zuul-jobs refs/changes/49/984749/1"}}},"commit":{"parents":[{"commit":"2b2c148b195f49b1ca5c063eebe1fc3f9aef4d13","subject":"Merge \"Drop legacy bionic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/2b2c148b195f49b1ca5c063eebe1fc3f9aef4d13"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-15 13:14:22.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-04-15 13:23:33.000000000","tz":120},"subject":"Add py310 job to openstack-python3-jobs for 2023.1","message":"Add py310 job to openstack-python3-jobs for 2023.1\n\nPython 3.10 was the supported maximum python version in 2023.1 Antelope\nrelease [1], but somehow the stable/unmaintained branch filter was not\nadded to openstack-python3-jobs templates.\n\nNote that openstack-python3-antelope-jobs job template exists, but that\nis not used, because that cycle was when the new openstack-python3-jobs\ntemplate was introduced and openstack-python3-zed-jobs were replaced by\ngenerated patches to this new template.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2023.1.html\n\nChange-Id: I503dbda1bbea1e2fecbafdd02b2e36bd254cc2f7\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6ee11fc0f01c3e3828886645509bbbd36fe613b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-zuul-jobs/commit/6ee11fc0f01c3e3828886645509bbbd36fe613b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
