)]}'
{"id":"openstack%2Fopenstackclient~691750","triplet_id":"openstack%2Fopenstackclient~master~I0fbba2baf021e666cba6a13c38e6152216ac5d7d","project":"openstack/openstackclient","branch":"master","topic":"openstackclient-plugin-jobs_docs","hashtags":[],"change_id":"I0fbba2baf021e666cba6a13c38e6152216ac5d7d","subject":"Copy in python-openstackclient docs","status":"ABANDONED","created":"2019-10-28 20:37:04.000000000","updated":"2020-03-27 22:09:49.000000000","total_comment_count":171,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f455e427e86fdb02a9b1999c5a88d003398bf9ae","_number":691750,"virtual_id_number":691750,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-28 20:54:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2019-11-29 14:52:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":-1,"date":"2019-11-29 15:40:05.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 20:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-29 14:52:33.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"75097383d335046ff9f603d7b54c2e97565ae55a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 20:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd731b81d30a112f4d39394622ed34f0863d2d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:37:18.000000000","message":"Patch Set 1: Verified-1\n\n(171 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstackclient on branch master.  The error was:\n\n  Unknown project openstack/python-gnocchiclient\n\nThe error appears in the following job stanza:\n\n  job:\n      name: openstackclient-check-plugins\n      parent: unittests\n      run: playbooks/openstackclient-check-plugins/run.yaml\n      timeout: 1800\n      required-projects:\n        - openstack/openstackclient\n        - openstack/osc-placement\n        - x/pbrx\n        - openstack/python-openstackclient\n        - openstack/python-aodhclient\n        - openstack/python-barbicanclient\n        - openstack/python-cloudkittyclient\n        - openstack/python-congressclient\n        - openstack/python-designateclient\n        - openstack/python-gnocchiclient\n        - openstack/python-heatclient\n        - openstack/python-ironic-inspector-client\n        - openstack/python-ironicclient\n        - openstack/python-karborclient\n        - openstack/python-mistralclient\n        - openstack/python-muranoclient\n        - openstack/python-neutronclient\n        - openstack/python-octaviaclient\n        - openstack/python-pankoclient\n        - openstack/python-qinlingclient\n        - x/python-rsdclient\n        - openstack/python-saharaclient\n        - openstack/python-searchlightclient\n        - openstack/python-senlinclient\n        - openstack/python-tripleoclient\n        - openstack/python-troveclient\n        - openstack/python-vitrageclient\n        - openstack/python-watcherclient\n        - openstack/python-zaqarclient\n        - openstack/python-zunclient\n        - openstack/requirements\n\n  in \"openstack/openstackclient/.zuul.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"3092eb70db39b226df48dda9689218036481e083","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-28 20:46:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9a4e8de32bc312a6c9a04274ebc1d98907eb284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 20:54:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstackclient-check-plugins https://zuul.opendev.org/t/openstack/build/73ccdd11369e42549de3dc439aca5bc3 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304ce930246642a3889ecda7e06f65ac : SUCCESS in 6m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce39e21b7c5d45bdbb074ebe80477f54 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d73708085ba643df8311fdee46e8d994 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc47055d2f2b40708ab9c02d5c5b330f : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cdd6296614b4d4aa73abfe081a56cca : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/755388e126074e30ad10d36d50e52532 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"70004b4d826489f4214a6ca49125fbaa73695256","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-29 14:52:33.000000000","message":"Patch Set 2: Code-Review+1\n\nAh, I didn\u0027t know there was an effort to do this. The infrastructure side of this is sane so LGTM","accounts_in_message":[],"_revision_number":2},{"id":"9ded8065047e94362cad8e2219f5ed9f01f591d1","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-29 15:40:05.000000000","message":"Patch Set 2: Workflow-1\n\nSorry, I must have discussed it with Dean in IRC but didn\u0027t update the patch. The  \"effort to do this\" was a bit premature; at the very least we need to wait until the python-openstackclient side is done being cut over to all autogenerated docs. The last couple of those are on hold, so we should wait on this too, and revisit it later.","accounts_in_message":[],"_revision_number":2},{"id":"457929a2f8e172384d6e3c2bd09890468e018ac6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-03-27 22:09:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7aea8e8e60f172ccbed1f8c565e70814c877f94","revisions":{"7fa0405f347b31ebff450f28ec593996eb4f4015":{"kind":"REWORK","_number":1,"created":"2019-10-28 20:37:04.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/50/691750/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstackclient","ref":"refs/changes/50/691750/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/1"}}},"commit":{"parents":[{"commit":"a8629fd4c1e431162c8b30d54918d02196869024","subject":"Update doc index links (for storyboard etc)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstackclient/commit/a8629fd4c1e431162c8b30d54918d02196869024"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-28 18:27:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-28 19:29:25.000000000","tz":-300},"subject":"Copy in python-openstackclient docs","message":"Copy in python-openstackclient docs\n\nThis populates the project\u0027s docs with (almost) an exact copy of what\u0027s\nin python-openstackclient at commit\n977b0c8591e40d3f59bbc1af9adc75f2c2edb694.\n\nThe primary purpose of this is to allow us to (in a subsequent patch)\nbeef up the openstackclient-plugin-jobs template so that plugin projects\ncan\u0027t merge broken option helps anymore.\n\nNote that we\u0027re still not publishing these docs. Before that happens, it\nshould undergo a read-through to sanity check that everything is\nappropriately non-python-openstackclient-specific (particularly the\ncontributor/ section). If/once we do start publishing it, we should\nremove the docs from python-openstackclient (adding redirects) so they\ndon\u0027t have to be maintained in both places.\n\nChange-Id: I0fbba2baf021e666cba6a13c38e6152216ac5d7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstackclient/commit/7fa0405f347b31ebff450f28ec593996eb4f4015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstackclient/commit/7fa0405f347b31ebff450f28ec593996eb4f4015"}]},"branch":"refs/heads/master"},"c7aea8e8e60f172ccbed1f8c565e70814c877f94":{"kind":"REWORK","_number":2,"created":"2019-10-28 20:46:15.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/50/691750/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstackclient","ref":"refs/changes/50/691750/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstackclient refs/changes/50/691750/2"}}},"commit":{"parents":[{"commit":"a8629fd4c1e431162c8b30d54918d02196869024","subject":"Update doc index links (for storyboard etc)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstackclient/commit/a8629fd4c1e431162c8b30d54918d02196869024"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-28 18:27:46.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-28 20:45:56.000000000","tz":-300},"subject":"Copy in python-openstackclient docs","message":"Copy in python-openstackclient docs\n\nThis populates the project\u0027s docs with (almost) an exact copy of what\u0027s\nin python-openstackclient at commit\n977b0c8591e40d3f59bbc1af9adc75f2c2edb694.\n\nThe primary purpose of this is to allow us to (in a subsequent patch)\nbeef up the openstackclient-plugin-jobs template so that plugin projects\ncan\u0027t merge broken option helps anymore.\n\nNote that we\u0027re still not publishing these docs. Before that happens, it\nshould undergo a read-through to sanity check that everything is\nappropriately non-python-openstackclient-specific (particularly the\ncontributor/ section). If/once we do start publishing it, we should\nremove the docs from python-openstackclient (adding redirects) so they\ndon\u0027t have to be maintained in both places.\n\nChange-Id: I0fbba2baf021e666cba6a13c38e6152216ac5d7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstackclient/commit/c7aea8e8e60f172ccbed1f8c565e70814c877f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstackclient/commit/c7aea8e8e60f172ccbed1f8c565e70814c877f94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
