)]}'
{"id":"openstack%2Fopenstacksdk~1001800","triplet_id":"openstack%2Fopenstacksdk~master~I4809866ef8b0b7840728853e826f9aee2944e949","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-22 14:37:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4809866ef8b0b7840728853e826f9aee2944e949","subject":"tests: Add some missing asserts, type hints","status":"MERGED","created":"2026-08-21 12:55:25.000000000","updated":"2026-08-22 23:38:11.000000000","submitted":"2026-08-22 23:38:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001800","meta_rev_id":"d492bbf5d7e73dd6018b94cb9436f10e481f6f9a","_number":1001800,"virtual_id_number":1001800,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-22 23:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-22 09:59:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-08-22 09:59:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-21 14:43:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0402c235967abf18bfc953f0517fd8013633462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-21 12:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a86b6e4181f75968e36212d8fb49ac7ce9040b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 14:43:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a451b3bf9ec54b1fa69c99082b3218a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d61bb39cd24d0181c575d6ee5a1566 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e99c1e3ea91a4b7394f5ecd664fdc09d : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cd2cab307814bada4484d238a3c6abc : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b05f7156f7ce47d1ad3c71a048bf43fa : SUCCESS in 3m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7563c26d7f1447778f4488af593dddf1 : SUCCESS in 6m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/9077fab1118549729f54ea24b7338387 : SUCCESS in 8m 30s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/26f0fc13ba2c4b439495bc141e72c8be : SUCCESS in 1h 10m 00s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cd04102038094491953178e6442660b7 : SUCCESS in 4m 29s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0a61df71a4154880bbda4d15d6391027 : SUCCESS in 3m 42s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4801657b7fb94c2fa6654b5b9402015f : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1bb72c959947bea9d4a2ef8ffb1a84 : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfbb11876bb444c6a327ef19f831fc1f : SUCCESS in 4m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fda3a6a1ba049cc89a02dca92aefb5e : SUCCESS in 1h 02m 28s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/ff9992e44bed42a680a922dad8a5536a : SUCCESS in 58m 26s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/3136b22aa47f44168aa4bec356bd9d2c : FAILURE in 1h 38m 44s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e2cb32a6220e40e291e32f9251476c5d : FAILURE in 1h 00m 17s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e31ebd56167c46179e1bfc2a5c465ba7 : SUCCESS in 37m 34s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/61bc1e30f0614d8889ba8c477d4fb180 : SUCCESS in 28m 37s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/848e137d1f5d433ba1209fe7b1f45527 : SUCCESS in 31m 02s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ef3e6d2fa45e4b33b8ba6a38e0b76d67 : SUCCESS in 28m 40s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/dc044d79d4874e53bfdc335b4ea164de : SUCCESS in 24m 56s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4ce8c140599543fda196b361758dc50c : SUCCESS in 1h 06m 18s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/7bbb5fba0f1d464e871a666672c0a996 : SUCCESS in 1h 06m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96754cb92d9ba08e234f668ee05d298990c82329","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-22 09:59:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c11457cdb6002ea41672c6f392f913072c3b8be5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 12:13:42.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/openstacksdk/+/1001799 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"e6bbad6d748b851cad530e5c31f15dc6e17a5bbd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-22 14:37:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16f19b44337a94e5e2f3c15055fdb60540e3a01c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 15:59:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5b60a623df41e6a18c2ce7e7f85900\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/990988ceb05e4275b6705204601ade5e : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5acc46718a624e87a0428aef12d12aae : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ccded7d640a48f0847842cf0039e376 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/314bfe69865c4470b2944a986fc333d2 : SUCCESS in 6m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82422e04f6544deebde05b574037ae33 : SUCCESS in 3m 54s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/66796d3641b6437f94678e5f3eb5c71f : SUCCESS in 7m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a3e27cd0149f43ff9db1be406d0f8c05 : SUCCESS in 30m 50s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/365f6a54613b48989cec62c0a838ca52 : SUCCESS in 3m 43s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ea986e37829d48a4a39110c1db22fdd2 : SUCCESS in 4m 57s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/77713c7b11ea4e5abc05e5b369ca7585 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/582f8583815e4caaa46fcaa1e0fc5886 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ebcc7d6938745778cb4d2aa87a2c4fa : SUCCESS in 5m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42bfa7b157e445ab88ebea7ace6a458b : SUCCESS in 32m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/082152d93c1844618235b84029ee60aa : SUCCESS in 1h 17m 11s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a4d6878e75d1471286bab9185ca2057a : FAILURE in 1h 10m 00s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/167d962300a54778bd4b3b8a33d3a1ad : FAILURE in 35m 55s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/289033600a7749aeaa973541825b3568 : SUCCESS in 32m 21s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/b47db111e1c34316ad1c973a7973c02b : SUCCESS in 28m 27s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/2e0093750a1e4c499dfaee96bba81751 : SUCCESS in 30m 28s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/9854d428e169452882be9992733ea330 : SUCCESS in 28m 56s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b5596191b04e477291b3859bc199da26 : SUCCESS in 39m 25s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/26db2b22305d4fb798aa839ec3426721 : SUCCESS in 1h 12m 32s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/4f15a0eb5821454a91ab7a0d2665671d : SUCCESS in 1h 15m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b73da092d2d51bff8ea7177f73ea1cdfb0bfea8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 22:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"52769d9f52365af76d825c04b33a33d1ae434634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 23:38:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd280a6d6b5145b7a5d28ff32689bfbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c471a03ae5f34d0b8a9c893b1364e4d4 : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a279bda4e8147a1aa27aae3bdb3ffb7 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ae45d51724c4d72b9da0049170d141f : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d4fdcb78fa440dc866253d1c65781a4 : SUCCESS in 8m 02s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8b7458730b96488db476c8050d783f86 : SUCCESS in 1h 00m 56s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/36febf7416b24a6685094411ad878b46 : SUCCESS in 3m 40s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f790e872d2e24c2786e09291f47c8c4a : SUCCESS in 4m 27s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b5ba24395efc4e00b3dbb85a0fcceedb : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21efa2231125462daae2b517d44e2734 : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8e2d9fc8948487e8f666ada3cb87be8 : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6be14477504c4ff8a2372c08766600c7 : SUCCESS in 56m 00s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/802cd597752843f09f3d788ff37195c1 : SUCCESS in 1h 23m 22s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/afa65446613d45bca7825bf9bd29d316 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d492bbf5d7e73dd6018b94cb9436f10e481f6f9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 23:38:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5c12de311ee0b3993215fe3b6d6b4e5f925e694","revisions":{"b5c12de311ee0b3993215fe3b6d6b4e5f925e694":{"kind":"REWORK","_number":1,"created":"2026-08-21 12:55:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/00/1001800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/00/1001800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/00/1001800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/00/1001800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/00/1001800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/00/1001800/1"}}},"commit":{"parents":[{"commit":"f93494ab36e5a3421fb7412bb965de838ad14570","subject":"tests: Use a module-level base class in resource action tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f93494ab36e5a3421fb7412bb965de838ad14570"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-18 09:50:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-21 12:44:33.000000000","tz":60},"subject":"tests: Add some missing asserts, type hints","message":"tests: Add some missing asserts, type hints\n\nGive mypy a helping hand.\n\nChange-Id: I4809866ef8b0b7840728853e826f9aee2944e949\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b5c12de311ee0b3993215fe3b6d6b4e5f925e694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b5c12de311ee0b3993215fe3b6d6b4e5f925e694"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
