)]}'
{"id":"openstack%2Fopenstacksdk~1003356","triplet_id":"openstack%2Fopenstacksdk~master~I1f8dc4dd3f96894ff5678ec7755d67b57eff996a","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-01 21:03:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-01 23:26:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f8dc4dd3f96894ff5678ec7755d67b57eff996a","subject":"Fix typo","status":"MERGED","created":"2026-09-01 19:01:56.000000000","updated":"2026-09-01 23:28:51.000000000","submitted":"2026-09-01 23:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003356","meta_rev_id":"ffc06aee4c4dac48add297a0d9bff676250e9397","_number":1003356,"virtual_id_number":1003356,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 23:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-01 21:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-01 21:03:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 19:06:34.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-01 20:50:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b55efbd25ad83eb6ae65d442a34a4a3ad40c2a15","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e5c2b1fe1de80a23c19cdbc571f0c22266d665b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:06:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7198dbc326b2d1d17bbc8d0fdc8a26b514e0a17d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:06:34.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"fa62c492014cad43eff9c243d93b53808770bb84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 20:50:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed520854b6e745a29638f6fa85bfcc0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3d37f3b6a7476c892ce7cd6ec41c71 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/188fe6186e154387bfbab0765f829c03 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f08f488b81647ee86104893980e1cd9 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e7328734a4f842fb916f07bc9d40134f : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/17f5d8c3962942518e714f912fae8698 : SUCCESS in 6m 36s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/98f8fe43f2aa41868635592406f0eab6 : SUCCESS in 7m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c3478b2661ae4bdbb55d3ee57afba4ce : SUCCESS in 1h 05m 48s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cea74f568b9c4688a292486ed833ef0e : SUCCESS in 3m 32s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/df434fc687e2484ba1266179ea71abaf : SUCCESS in 3m 10s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/659c2007a3f84eb2933243d626e1777c : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004d68ca2f3c4d6a98aa2a7e9fe88f84 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6d6d656c9e74e939cf2f5f81acec3d4 : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0fafb36e0594c18b901aa683be158dd : SUCCESS in 55m 38s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/47e2e2b0e0984fa0a2c53ad57b55558c : SUCCESS in 1h 25m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f876ec15b9ce4adaac4b3e437ee36d1d : FAILURE in 1h 40m 35s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/aff812c5c90a4d11b7a8f333b264517d : FAILURE in 28m 46s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bbe035684ccd4a1cb6439260ced86b41 : SUCCESS in 34m 07s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/747cc8e863af4e0ea5912da6a0fbed6f : SUCCESS in 29m 30s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/cb325dff286c4624ab95f6f6c5f3dc69 : SUCCESS in 21m 22s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/cb37956807a94678a2134b5dc6b03e12 : SUCCESS in 29m 24s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/c324a547d81e4747a1e65f81d81ad902 : SUCCESS in 44m 00s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6f84c0d76a1e49fdae9c10c554507c23 : SUCCESS in 1h 15m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3d57299a5da54eaa856b62e26f438ed0 : SUCCESS in 1h 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9459ec63a63cc58c4f19dc17276464c913ef975a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-01 21:03:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1616f89007520cbf34d7a9cb447aad97bc239f4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:03:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26426eecebdf5ce19793e159cd6be5ef2646b204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 23:26:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00592fb5d1744afebcb9a5692821a566\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c950b59681f4b5a9ad828ca9858dfba : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38fb0eeda0474fb48d5f40626b4e79ee : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8bfb69ef90444156873edd096037bd59 : SUCCESS in 3m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f5af0fef64ac4ac5bf5b94c6436c3acb : SUCCESS in 6m 33s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3c85e3dd0140432497a23e9184b36210 : SUCCESS in 41m 01s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/53522cfd1cdc4fea9ba23e3747116835 : SUCCESS in 3m 42s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/67931c5882c0425ab2e66d25d81f29c4 : SUCCESS in 5m 23s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/77c12fea13cc491e8f4a6e80faeb265c : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4abc15006244c84a7c5548b0e083ad0 : SUCCESS in 15m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2ba5b5f86554bc5b3a7c7c2bb7e3103 : SUCCESS in 4m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc6c1c6b2e64cdc8a519fb38ad3eafc : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/578657ed70d94aa8a1dc8caa62b9a85d : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/926c8815b2b84c6da8f69142ce182f48 : SUCCESS in 17m 57s","accounts_in_message":[],"_revision_number":2},{"id":"4ca3e7f355e9184d36f460ed92ed32292713ad27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 23:26:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ffc06aee4c4dac48add297a0d9bff676250e9397","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 23:28:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a35285827f454aa89bef948955611ab5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ece248004d49e4a787f7f12f3279b9 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4830e2f9131143310976c3fafd66fbaec63b3620","revisions":{"2e41c1beae395bd851e0e0d04991e51faa41e528":{"kind":"REWORK","_number":1,"created":"2026-09-01 19:01:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/1003356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/56/1003356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/1"}}},"commit":{"parents":[{"commit":"045d216812d7fcac4f60742c8cef5fad1be6f522","subject":"Add a way to discard the authentication cached for a cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/045d216812d7fcac4f60742c8cef5fad1be6f522"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:01:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:01:56.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I1f8dc4dd3f96894ff5678ec7755d67b57eff996a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2e41c1beae395bd851e0e0d04991e51faa41e528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2e41c1beae395bd851e0e0d04991e51faa41e528"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"4830e2f9131143310976c3fafd66fbaec63b3620":{"kind":"REWORK","_number":2,"created":"2026-09-01 19:06:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/1003356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/56/1003356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/56/1003356/2"}}},"commit":{"parents":[{"commit":"045d216812d7fcac4f60742c8cef5fad1be6f522","subject":"Add a way to discard the authentication cached for a cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/045d216812d7fcac4f60742c8cef5fad1be6f522"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:01:56.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:06:22.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I1f8dc4dd3f96894ff5678ec7755d67b57eff996a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4830e2f9131143310976c3fafd66fbaec63b3620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/4830e2f9131143310976c3fafd66fbaec63b3620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
