)]}'
{"id":"openstack%2Fopenstacksdk~1003636","triplet_id":"openstack%2Fopenstacksdk~master~Ic29bff0601bc184abd4de54c76be358a7c6dfe09","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-04 16:16:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"39399":{"account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"last_update":"2026-09-04 19:18:31.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-04 17:02:03.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"Ic29bff0601bc184abd4de54c76be358a7c6dfe09","subject":"stats: Fix prometheus histogram units","status":"MERGED","created":"2026-09-03 09:23:53.000000000","updated":"2026-09-04 19:19:39.000000000","submitted":"2026-09-04 19:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003636","meta_rev_id":"115d4a3d6352a6f35397631bd3b1147d5ae5bfa6","_number":1003636,"virtual_id_number":1003636,"owner":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-04 19:18:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-04 16:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-04 17:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-09-04 17:02:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 10:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 13:30:03.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-04 11:28:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"947bd5efdf64035965080bf89dcda882007e54f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-03 09:23:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"008cfa40101f866de004675c6338c024d5bbf7fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 10:52:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c128d93e464b8abff9304d4be5362d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54c65ba312242c18f7f73688bcd8179 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b905969655f747bea60794e60e07c4dc : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90523ad0dc94416bb5573bf4d49cdb08 : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/225f3a6c2ee94dd495fbc50c1e6b9425 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/26e6712a979e4032b5d6d34564165999 : SUCCESS in 7m 20s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/2f623d87c246413d85208713a265696e : SUCCESS in 10m 05s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c0aa560264984412b19539c42e2762f6 : SUCCESS in 1h 05m 03s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f42b84cf3185494abb812140f045b69d : SUCCESS in 5m 10s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f3175e351ac94c4bae446adb1da3dd81 : SUCCESS in 4m 30s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/537daf6cd0014e42842dbd94da20cf83 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c81ab8ee8bc4537bc836b6884b72a14 : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba98a3099e46497888353414c1fdc985 : SUCCESS in 4m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff142700bc594f5091f96f7208b3d6b2 : SUCCESS in 1h 02m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/25b9c03a0f244f32880df9832a712e5a : SUCCESS in 1h 22m 33s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/4f63ac5623724828ace6b66ceddb2862 : FAILURE in 1h 19m 43s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a2584f8acfb3404892fc173a637a218c : FAILURE in 55m 16s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/0691d5aea2094c22bce61f89b0dc8565 : SUCCESS in 36m 11s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f575fee8700d454887e6d8c5eef9acf2 : SUCCESS in 31m 21s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/33a4d93cb97f457ca2f3cb49d519a451 : SUCCESS in 22m 43s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/1a15fe7293b44e0e913b440c4feb1177 : SUCCESS in 31m 17s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/49f043ae025d405c8286080c627eaa3f : SUCCESS in 41m 17s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b0580ce30a6846db9bf5b0b8548fa4c1 : SUCCESS in 1h 09m 59s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ef57ec8045be4ce4b391d1fe9e037ae0 : SUCCESS in 1h 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"307928ffb899c9864a5e19ae43024e5319cd127f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:30:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"061ec42e2e09046012a222e8cce0aceefa3994a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-04 04:59:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7ef321d7a8786c122a87786dea2e478e9c54f14b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-04 05:45:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a12a41a60dc769296a5eb6834e1163d58c51a55","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-04 05:53:49.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"24a2cc6b891808a5510e555895d4b668859b375c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 07:06:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f430c74920c047c0908ebe605194c7f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4317cc999e5e49cbb0805b3faa46c5a2 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb5b2999322d44fe834194405c17b46a : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cefc92829cf74693a48bb2223980ee23 : SUCCESS in 5m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c811851e7a49472c94da1f9b0d3d3086 : SUCCESS in 4m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0416d98e6da14e0da9675c24aaee71ab : SUCCESS in 6m 38s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4ac913c9df144e73ae2fef6f8bbbbb4a : SUCCESS in 4m 44s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/158bade1f4134893885e5c82ac33bd33 : SUCCESS in 49m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/643e49c32aec404bb8cf95e17a489557 : SUCCESS in 4m 13s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0edb8baf28324ba08bd3df47e028a213 : SUCCESS in 3m 43s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/692fa5dca5e0484eafec83b5ab0bdc7f : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdbbc2410713447f801fbb6cef0f1b3e : SUCCESS in 13m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1752b68e199c45558c74978645a0a047 : SUCCESS in 5m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d46cca5700ce408fb766d3a5c46d51cf : SUCCESS in 1h 03m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3767d569af754fb8ac0c163bc0534f75 : SUCCESS in 51m 38s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/733dc7675014480884cd94fd91790bb1 : FAILURE in 1h 18m 36s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ff51cdd24d4a4b749604a1de456097a5 : FAILURE in 58m 09s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/90bd5553ec8c4b228bf4038c70358605 : SUCCESS in 21m 56s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/618d5f53aec64b08ae2c8b238b8a5030 : SUCCESS in 29m 47s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/4fe7ebd643a94935a424d9f37c8f84e7 : SUCCESS in 15m 16s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/cbcc6f059c1f4ae899d76c58b400d54e : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/65b075b1461f4219b69c41e991aa581e : SUCCESS in 42m 33s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e832992d71654dc3841b0dc95dd428ec : SUCCESS in 50m 53s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/87b77b8442b04fc98f220d50d3bf28a2 : SUCCESS in 1h 13m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21112fc4627e48531a6f012db556912bd98eedc1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-04 11:28:15.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"493c4f746e673bed88ddc6e7f3d824c6424efda6","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 14:29:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3567e4f92fa8a16572ae189bc5a03fb18a517b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-04 16:07:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"8cec21eaf5fc679a1cd925fea74748e54a56a90b","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-04 16:15:34.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5e3398e2b92ae7f310cc4a8524607d883bb48822","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-04 16:16:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dcb693c104d19c5a72bb0012042438b58c472bf6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-04 16:16:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"315db8f7f1ce1c73e92975c355248103cd5cf7cb","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 17:02:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"73a92e0ea72dd5ddcf5c427ef6612b9dfe0ec122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:38:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0261aa0ed0624b65971508bb9648c6ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365f9dba98c5403784f18a1de816c37b : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f018e4f159f5406b86c36b7962d00c1a : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6464e32aa8bb4e9ba0c0c4921d295398 : SUCCESS in 6m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7fc1ffd027ad48e6b0337e40fb925d90 : SUCCESS in 4m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34e5e6625fd54b9da2616f53063b82a4 : SUCCESS in 8m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c6bfbe2a47b6439c86e805fd14bf6a49 : SUCCESS in 6m 32s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b97e3ddf532e4961b614ade939763ee3 : SUCCESS in 1h 02m 09s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fa1e386e82bb4f169ea61b6063266b01 : SUCCESS in 4m 44s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6ee7c39b82c54e72abf1d710a555cdd8 : SUCCESS in 4m 56s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/e2b8e009e23c4c24a53d30395b1ec249 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92e4783393fa41c595a0fd500380ac8b : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4967cbd876094b04bffc9093c3f9d197 : SUCCESS in 6m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c53a2ea49bc4ef498fb027cad5ed2d8 : SUCCESS in 27m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/50a3733ff55b4ad086962f1f0708e083 : SUCCESS in 50m 31s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/34de4308f7af400ab6cfe86fa15b8796 : FAILURE in 1h 24m 20s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b48702e48d0b4aaf92b034b83a266929 : FAILURE in 55m 29s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f35756386a2e4dcd81e42776a79eef0e : SUCCESS in 35m 53s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f5ccb888406c428889ba5e3640e92d49 : SUCCESS in 33m 13s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/bb6867f4a295483cb8ceb2065b2077f6 : SUCCESS in 28m 51s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/36bb72950059477789bb47d0a2dc3bc4 : SUCCESS in 25m 09s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b271b7600b0e4b0ea9646959169b6cce : SUCCESS in 44m 18s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3ee6c4602e004a61859fbd8ae65e7b2e : SUCCESS in 1h 11m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/438bc918201743c6884286cdc8b1edd2 : SUCCESS in 1h 03m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"66110b8bde8c067d7f3fb13f5f10bf3958db80aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:38:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5afe9db79136485fcfe4b5b0be329b6748333e7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:18:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f44347c27264d428d3b04428efa7e57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c79fd51517a4109ae20e59760f902fa : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b371fca764eb4e279581e761ebb27bf3 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/341c01eca798433184f2ff71f8678315 : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/21985aff89014e5fb9eaebdd02447906 : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/032dfb286a06451b95ee8a06080e70f9 : SUCCESS in 1h 01m 34s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2e6dcdd433e0465880b05616f07d0413 : SUCCESS in 4m 00s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cb762dc0d4054d789fdfa93e2c8f21ec : SUCCESS in 4m 19s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/7a5601ae3bfe4bf38f799552284679e1 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d5b5f814915477884d55e73b2ed7733 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2b4f21b324943848dab5d3efb967d4e : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86dc0a8cd56d4b0f9895ffb3d59b4ff7 : SUCCESS in 1h 01m 07s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/537d9968b94344679d011876fabe74f4 : SUCCESS in 1h 29m 53s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/a3b52dec2c1646ba8ce2e1648730ee50 : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":4},{"id":"40df7eb6cacc7391b8681bfa02d103b82b1be44a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:18:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"115d4a3d6352a6f35397631bd3b1147d5ae5bfa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:19:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a524e31a41483cab0e56a871d9e98a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d030dc343304c808bbc35d74c894987 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d07f3585294e3b980d2c874ed16186f92e67afc9","revisions":{"67cb9da81deef7d3b2e8b71f0d101c24e0e57e66":{"kind":"REWORK","_number":1,"created":"2026-09-03 09:23:53.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/36/1003636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/36/1003636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/1"}}},"commit":{"parents":[{"commit":"045d216812d7fcac4f60742c8cef5fad1be6f522","subject":"Add a way to discard the authentication cached for a cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/045d216812d7fcac4f60742c8cef5fad1be6f522"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-03 08:57:18.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-03 08:57:18.000000000","tz":420},"subject":"stats: Fix prometheus histogram units and add custom bucket support","message":"stats: Fix prometheus histogram units and add custom bucket support\n\nRecord response time in seconds instead of milliseconds to match\ndefault prometheus_client histogram bucket boundaries.\nAdd metrics.prometheus.buckets option in clouds.yaml for\nconfiguring custom histogram bucket boundaries. When not specified,\nthe prometheus_client library defaults are used.\n\nCloses-Bug: #2069431\nChange-Id: Ic29bff0601bc184abd4de54c76be358a7c6dfe09\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/67cb9da81deef7d3b2e8b71f0d101c24e0e57e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/67cb9da81deef7d3b2e8b71f0d101c24e0e57e66"}]},"branch":"refs/heads/master"},"e6f15428be5030e80a9b324febc2d02a98c548f4":{"kind":"REWORK","_number":2,"created":"2026-09-04 04:59:03.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/36/1003636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/36/1003636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/2"}}},"commit":{"parents":[{"commit":"045d216812d7fcac4f60742c8cef5fad1be6f522","subject":"Add a way to discard the authentication cached for a cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/045d216812d7fcac4f60742c8cef5fad1be6f522"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-03 08:57:18.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-04 04:57:53.000000000","tz":420},"subject":"stats: Fix prometheus histogram units","message":"stats: Fix prometheus histogram units\n\nRecord response time in seconds instead of milliseconds to match\ndefault prometheus_client histogram bucket boundaries.\n\nCloses-Bug: #2069431\nChange-Id: Ic29bff0601bc184abd4de54c76be358a7c6dfe09\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e6f15428be5030e80a9b324febc2d02a98c548f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e6f15428be5030e80a9b324febc2d02a98c548f4"}]},"branch":"refs/heads/master"},"fd6cb07844e6fbb1f8cf34058e5e8f4bd87243e2":{"kind":"REWORK","_number":3,"created":"2026-09-04 05:45:52.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/36/1003636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/36/1003636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/3"}}},"commit":{"parents":[{"commit":"045d216812d7fcac4f60742c8cef5fad1be6f522","subject":"Add a way to discard the authentication cached for a cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/045d216812d7fcac4f60742c8cef5fad1be6f522"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-03 08:57:18.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-04 05:45:29.000000000","tz":420},"subject":"stats: Fix prometheus histogram units","message":"stats: Fix prometheus histogram units\n\nRecord response time in seconds instead of milliseconds to match\ndefault prometheus_client histogram bucket boundaries.\n\nCloses-Bug: #2069431\nChange-Id: Ic29bff0601bc184abd4de54c76be358a7c6dfe09\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fd6cb07844e6fbb1f8cf34058e5e8f4bd87243e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fd6cb07844e6fbb1f8cf34058e5e8f4bd87243e2"}]},"branch":"refs/heads/master"},"d07f3585294e3b980d2c874ed16186f92e67afc9":{"kind":"REWORK","_number":4,"created":"2026-09-04 16:07:33.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/36/1003636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/36/1003636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/36/1003636/4"}}},"commit":{"parents":[{"commit":"045d216812d7fcac4f60742c8cef5fad1be6f522","subject":"Add a way to discard the authentication cached for a cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/045d216812d7fcac4f60742c8cef5fad1be6f522"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-03 08:57:18.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-04 16:07:11.000000000","tz":420},"subject":"stats: Fix prometheus histogram units","message":"stats: Fix prometheus histogram units\n\nRecord response time in seconds instead of milliseconds to match\ndefault prometheus_client histogram bucket boundaries.\n\nCloses-Bug: #2069431\nChange-Id: Ic29bff0601bc184abd4de54c76be358a7c6dfe09\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d07f3585294e3b980d2c874ed16186f92e67afc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d07f3585294e3b980d2c874ed16186f92e67afc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
