)]}'
{"id":"openstack%2Fopenstacksdk~1003742","triplet_id":"openstack%2Fopenstacksdk~stable%2F2026.2~I9a9e1fade0d10c3d11a8c14751836803f6a538e3","project":"openstack/openstacksdk","branch":"stable/2026.2","topic":"create-2026.2","hashtags":[],"change_id":"I9a9e1fade0d10c3d11a8c14751836803f6a538e3","subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","status":"MERGED","created":"2026-09-03 11:06:58.000000000","updated":"2026-09-03 15:08:28.000000000","submitted":"2026-09-03 15:02:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003742-create-2026.2","meta_rev_id":"c4bdc132d8b1f539750960d55a1aa715e3615f61","_number":1003742,"virtual_id_number":1003742,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 15:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 12:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 12:23:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 12:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 12:23:25.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"3abeabc74278ef9fde7282308afc34038f91ec00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2026-09-03 11:06:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eab9d13714b6e4e5802f82c6ee1e16b486769b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17bd9d63b2954a3f80db00abf12aafe3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8dec7fbf5f34041b7c7391f25faeed5 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11b9a6e5e83a40309c4a23fa0ee2bc4b : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9547c975cc6a41f9b8e07fd7fed3cea1 : SUCCESS in 4m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b3625543b80e4f0292865ff4037cda77 : SUCCESS in 5m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dff174c1c41b4c4c9b0339eb1de2c3fd : SUCCESS in 7m 32s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/25aec814618e4420adc8b34c3f9f5780 : SUCCESS in 7m 36s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/384ddf65c5bb42e1b7dfd584e1fe39f9 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c548eb6443b4c83a3e07e68f6e882b9 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cb8f28345cc4b2aaf89ec708fe06805 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"902939130f1a7be3f4de4fae2c6493dc9da05157","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 12:23:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efdfa2b76a6069ab350524a68d0df7c9f339fd44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:23:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2987098efe1e53cb98b26d438a81d005f85ba2f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3693ad0d69748ee9d0052e14137723b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbccc06a652542d692426c4b622ef1fa : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed8bd4f16db14598bdde3b0faa3335f2 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5642de4cc8054325b46d21222b917834 : SUCCESS in 4m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/851d14d54980492e8c760be0aab2d463 : SUCCESS in 7m 03s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/122ad0947609428f856f61aae80425d7 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5041bfaf69b445989b541d26ffaa242d : SUCCESS in 17m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb9c20bb122b4981ae796f31a92c8c0e : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8524e1cfd4e9b9fae052e56201f1199929ffc5c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:02:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c4bdc132d8b1f539750960d55a1aa715e3615f61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:08:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619b2aa0ba1c4788ad4dcf9f0e32763f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf7d655ed6a4696bbb0558e31fb4051 : SUCCESS in 1m 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3545a8aca00b4f9ca1382cadbf9480a7 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ad23453742ffdae56ce7f3c8f6cb8948b4a3c00","revisions":{"3ad23453742ffdae56ce7f3c8f6cb8948b4a3c00":{"kind":"REWORK","_number":1,"created":"2026-09-03 11:06:58.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/42/1003742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/42/1003742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1003742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1003742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/42/1003742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/42/1003742/1"}}},"commit":{"parents":[{"commit":"d39e645bddcbecfdd9ed48f00970cb9c48118fc6","subject":"Update .gitreview for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/d39e645bddcbecfdd9ed48f00970cb9c48118fc6"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 11:06:56.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2026-09-03 11:06:56.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2026.2","message":"Update TOX_CONSTRAINTS_FILE for stable/2026.2\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2026.2 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I9a9e1fade0d10c3d11a8c14751836803f6a538e3\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3ad23453742ffdae56ce7f3c8f6cb8948b4a3c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/3ad23453742ffdae56ce7f3c8f6cb8948b4a3c00"}]},"branch":"refs/heads/stable/2026.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
