)]}'
{"id":"openstack%2Fopenstacksdk~1003768","triplet_id":"openstack%2Fopenstacksdk~master~If002b5af1a32ba407a77fcf47d6f8afab1b025dd","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"If002b5af1a32ba407a77fcf47d6f8afab1b025dd","subject":"config: Rewrap exception message","status":"MERGED","created":"2026-09-03 13:27:58.000000000","updated":"2026-09-03 16:43:50.000000000","submitted":"2026-09-03 16:42:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003768","meta_rev_id":"4bc431c5c5f6e3e3cdfcc08fc178d3a7cd1889e2","_number":1003768,"virtual_id_number":1003768,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 16:42:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 13:30:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 13:30:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:27:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2e8e086c98bae1110f0b8a130baf05f9e12f80e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9d900ab9b50ad4ea08a8bab7338d4f4a7942e9f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:30:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ee8ba11a1e8e07fa3f797941134931e5be876c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:27:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2e1d9ee36743fca5402ceccb7d2de3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c46d9f781584bd28b14f43ad7bed1cf : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c7140f982ad4efcaf2afbd1c984600c : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5442f6d406c643828fd17798cd36c7d1 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5054db555484ac798ae4ac7a8e4a585 : SUCCESS in 5m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2991f2ba5b214871b7800787e67b911d : SUCCESS in 7m 29s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/ccf3060394a141c6ad2be3e5a6855cbd : SUCCESS in 7m 38s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6a113f71e40f4ffeb06932715b097a9a : SUCCESS in 38m 33s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7dfad02f627b4bdab5a4d90837da6232 : SUCCESS in 4m 12s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e88f4022331945ad877f009b009d76f2 : SUCCESS in 5m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/7f10a288f68f4061a711a794ea2ce468 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1739f3d5e9674002956d4f7d9e3a57ed : SUCCESS in 14m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51189729d3ef4bcc862701b1acdc2404 : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a66867e957a04ddfa15078e22944b629 : SUCCESS in 1h 01m 22s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/da8417fe85604b8eb30ed678dcbc728a : SUCCESS in 1h 02m 01s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/208809f90a9947c1a008fa3367b2fe0f : FAILURE in 1h 27m 37s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/3d64c37109ca4bf4968ee97b701c11c1 : FAILURE in 53m 23s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/b8634062ea13459780f3136876b47e67 : SUCCESS in 21m 12s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/ad070e5635a64b59a9efe5949372c870 : SUCCESS in 32m 28s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/51ad204aaff34dcbb0bc98e5e0a1867c : SUCCESS in 29m 38s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/0d543381aff04d8fa71055533617a31a : SUCCESS in 35m 11s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e2c150db0b7f40d69b8f7ddc93a39062 : SUCCESS in 34m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8e0384c95f3d4ff9bf96d24dcfacfedb : SUCCESS in 56m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/32724595af7f4554ad858cfa07ea8a15 : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8bf4d4df0750d8a3633deca589b40d44b05f267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:27:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b94aef4b9a6593862cd2ec70f65f85e8ec2e4a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:42:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/751b91b8c30e461aa31b8837bc830bea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/084abb5c686744c9bf3b72c1e0a66508 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de1f88c7ac8c4defa69dee06d688316d : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e891d88bdd6e498db0f0688fb4b550e0 : SUCCESS in 2m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a756e5fc6f274e44ae0f80f1942192c1 : SUCCESS in 6m 25s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f3b16891ddc84f14b3ac64c48dac96e2 : SUCCESS in 1h 05m 51s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/59a659362e684731bcb3d299ece7a38e : SUCCESS in 4m 53s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6c782e15ad1d42f08587b533d70831cc : SUCCESS in 4m 05s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/e245e012e39045dc99286056e4ea29ef : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a045c826947f4716a70d76cbe5cf20c7 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/462fad4580b343d4bf4e967853046a52 : SUCCESS in 4m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b160603c61541e5ae98f30a8b119e46 : SUCCESS in 29m 59s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0d7bcfd02dc04297a471b6bd0fe854c8 : SUCCESS in 1h 13m 45s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/27465b67a399410c97a30b8b0240c3e7 : SUCCESS in 22m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f1efa63e91ae859f0f6003ef7a20b13df6a8228c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:42:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4bc431c5c5f6e3e3cdfcc08fc178d3a7cd1889e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:43:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/930b6e6cfe33414193fa3b0464a2f32a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b4190a57e941d8aba8bef765a2e812 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edfce0de27356b54ad7b99119c27dece406a9a06","revisions":{"edfce0de27356b54ad7b99119c27dece406a9a06":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:27:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/1003768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/68/1003768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1003768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1003768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/68/1003768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/68/1003768/1"}}},"commit":{"parents":[{"commit":"1d857a09d92608d8c4d6fb5d39235031e1f9f317","subject":"docs: Move MFA section under auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1d857a09d92608d8c4d6fb5d39235031e1f9f317"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-12 09:29:05.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 12:46:08.000000000","tz":60},"subject":"config: Rewrap exception message","message":"config: Rewrap exception message\n\nChange-Id: If002b5af1a32ba407a77fcf47d6f8afab1b025dd\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/edfce0de27356b54ad7b99119c27dece406a9a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/edfce0de27356b54ad7b99119c27dece406a9a06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
