)]}'
{"id":"openstack%2Fopenstacksdk~1003769","triplet_id":"openstack%2Fopenstacksdk~master~I61aeb4c6fc343a025d5696d27c663ff2f36b106b","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I61aeb4c6fc343a025d5696d27c663ff2f36b106b","subject":"config: Add missing docs for CloudRegion","status":"MERGED","created":"2026-09-03 13:27:58.000000000","updated":"2026-09-03 17:22:45.000000000","submitted":"2026-09-03 17:21:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003769","meta_rev_id":"10f8bf9f4d58a970acfa24e6ca20b6ccc3adf031","_number":1003769,"virtual_id_number":1003769,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 17:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-03 13:30:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-03 13:30:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1244397bc4e4ab1c6ea0fb71594be192e1e181f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bf750f726f7805117dbe02115d15aa95e15970a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:30:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"288d8c7cce294d3fc3cf219c6f3c81eb64208d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:42:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c564fc1683c24dcd9d55ebbc766a2b6a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4b75acfa5cb4bb4bee784b3a2d36ab5 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18d8563085184b7f97eea6e72b2579ed : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/22644131d08f4672993c470d9fa25d42 : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9bb1a1c4e174fd4b43bf5b89962d9aa : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2292ec176cba4ba2947ab89ea617e88c : SUCCESS in 7m 02s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b265f1b7754043be99536f43e02092c9 : SUCCESS in 7m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fe12e5fa78fa479e95d6fc384f0e160a : SUCCESS in 1h 01m 25s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9d4407b6db044774a15d04a78f0f4223 : SUCCESS in 4m 50s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5360247a3e3f426fa3f4633260f3e31a : SUCCESS in 6m 08s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/bedd7def64134cc1895c37978db2102b : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b786dc61d41d4a51b4a80012fbf1592c : SUCCESS in 15m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/965d51f169cb4cf6bca02b4ae7bf5cb3 : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c14724ced70642a899934d4cd1b563fc : SUCCESS in 1h 04m 30s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/385d2fbc13ac4609b188403263d05233 : SUCCESS in 1h 25m 13s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/6d97fbacd11244c5992082158fe8c158 : FAILURE in 1h 39m 41s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/5153eaef235a4b5bba096aabf990caf1 : FAILURE in 55m 58s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/f6e777429fdd408597111f8d04e53dda : SUCCESS in 37m 37s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/8a17a12fa1b0482f85a322c8eade0f8a : SUCCESS in 20m 20s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/332121f6794e4008b6997306b78bbfde : SUCCESS in 21m 20s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/bb9d592864f647889af5ebe5b6721865 : SUCCESS in 33m 51s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/51159a3ce3864184b8b546d990ad63e0 : SUCCESS in 43m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d9f3712d512b4fb496a654cdecf81295 : SUCCESS in 1h 15m 14s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/24cece27dc6b4ba39e4c6bfe82650583 : FAILURE in 1h 03m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3af193a4a764f287fe7231f9488a77f1afb94c39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:43:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d82f7fb990d17965b6d9b508d9b4a824b839338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:21:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5793ff9d26340d3bd6e450399c91f9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/920dc4220dac4cccb5bb6dabeb0cf0cc : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ceaad77c2a544a9b50fadac52fcc16b : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/712aaaafb99b4139adbf79348ce1f0b2 : SUCCESS in 6m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cd6d089a89fd42d4ad35d8fc2463f0f9 : SUCCESS in 8m 52s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/08062e90c5cf4f81b796f27e9fd84659 : SUCCESS in 1h 05m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9d2c98a6c5d84560b3b0587aa21a7682 : SUCCESS in 2m 53s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/45d84003cf8a4f60b393a7c3ba11fe03 : SUCCESS in 5m 04s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/7d91ecde499c4b5c96b4fa9b3a290a7c : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785e8ae849b149a5a086290da2cdc41d : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/345a748ccf644eb7b2958465d9be5e84 : SUCCESS in 3m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a496fdf1cec48169e83f75082b6f381 : SUCCESS in 58m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/68548cef9e354933b626bdda3b0363e2 : SUCCESS in 1h 31m 48s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/35628c25656c40e9a41ebad5e987399e : SUCCESS in 41m 36s","accounts_in_message":[],"_revision_number":1},{"id":"548266579b508cf0173c767f1de78db2d02f442a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:21:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"10f8bf9f4d58a970acfa24e6ca20b6ccc3adf031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:22:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b15e0fbada945b9b98a76e06e769bca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc954ec14aae43acaa790e8a7f3c3c87 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b207acd9c6776848fa498c5cf7b924514856df95","revisions":{"b207acd9c6776848fa498c5cf7b924514856df95":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:27:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/1003769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/69/1003769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/69/1003769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/69/1003769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/69/1003769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/69/1003769/1"}}},"commit":{"parents":[{"commit":"edfce0de27356b54ad7b99119c27dece406a9a06","subject":"config: Rewrap exception message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/edfce0de27356b54ad7b99119c27dece406a9a06"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-20 14:44:08.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 12:46:08.000000000","tz":60},"subject":"config: Add missing docs for CloudRegion","message":"config: Add missing docs for CloudRegion\n\nResolve a longstanding TODO. We omit docs for 3 options that we plan on\ndeprecating shortly.\n\nChange-Id: I61aeb4c6fc343a025d5696d27c663ff2f36b106b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nAssisted-by: Claude Opus 4.8 \u003cnoreply@anthropic.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b207acd9c6776848fa498c5cf7b924514856df95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b207acd9c6776848fa498c5cf7b924514856df95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
