)]}'
{"id":"openstack%2Fopenstacksdk~1003770","triplet_id":"openstack%2Fopenstacksdk~master~I4de8a2852782d70710911f46e330a74978590829","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 14:19:55.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I4de8a2852782d70710911f46e330a74978590829","subject":"config: Separate parsing of .client section","status":"MERGED","created":"2026-09-03 13:27:58.000000000","updated":"2026-09-07 15:55:05.000000000","submitted":"2026-09-07 15:53:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003770","meta_rev_id":"a8889666dd769fc7a84dd4e9e033d3c27045ec6f","_number":1003770,"virtual_id_number":1003770,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 15:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-04 14:19:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 14:19:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:12:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 14:19:27.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"8135382fe0872191777d6b384570e3af8ad4975b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b784d8f5841e3fc1f992c170e069d4659e5c9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:12:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb3598f3d1b44ab89b3b482d26937ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5382aa07f7764dce88fdc4450f8aa652 : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2680958bfe446dab11f6622a64a88ec : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67887affde894494884a2b1becf492c8 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0fc47b1a83e34e259ed100c30ef9c2f6 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dd88f493f0cf40a18d2db75649a90e42 : SUCCESS in 5m 22s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/34367387444343e09a642987765c15f7 : SUCCESS in 9m 31s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/39d5cfe625484af39033378536be24d1 : SUCCESS in 56m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7b26d7e9d0894638b64524bf8599e89d : SUCCESS in 7m 06s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ce88f66a652e40e7a3fe1972a830bd94 : SUCCESS in 5m 19s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/7d1169a2fb8f48ae96edbb043a5690e5 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147faf92a8284cdeb822b0b613c3f150 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61fa096cf51b42ba9979b9e69a9cfb09 : SUCCESS in 5m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a55c5a180464739a94a89e0c56dd208 : SUCCESS in 32m 15s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/4f4b0e6b8c464f729003962d794b6b99 : SUCCESS in 1h 13m 39s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/98bd854f2e07414593ecb95d13e284b0 : FAILURE in 1h 18m 14s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c7d6b99be8764f778928871503a27d80 : FAILURE in 27m 23s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/bf5835f444a1438d82fb16bfe2076334 : SUCCESS in 23m 44s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/e41b3383d36d4844976c10a6300f2a9a : SUCCESS in 17m 58s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/4d26d17eeec74955bee7c758b58cf6d2 : SUCCESS in 20m 31s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/2c0026cd48cc446cb1ce0b2312af9b40 : SUCCESS in 29m 29s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ac022b973a86405db7efbe78190e79bf : SUCCESS in 31m 22s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/44ee41e0c81d47999831ef46edc929e0 : SUCCESS in 41m 12s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/f7353e069b214e16b262c1a3788ba415 : SUCCESS in 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53891b744567d5059f2d984d317d447164a0a416","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 14:19:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4dd66940f3785835a54e152e1369b063050d7d71","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 14:19:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a21a99b01296cffee1944d56d75fa76b995e801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:21:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"573f25b1b41a952113946decdf857b92c8ddab8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:53:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff7cfbdec19240479ecea1491fc654c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e246786192104336b681aff2de5ea48e : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14b25409ba4e49049aba0a28d6b63071 : SUCCESS in 6m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2abed44f9d0b4c7db2c7ee1109e79b3b : SUCCESS in 4m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a68daf2a50fe48f1b49f059175986b2b : SUCCESS in 6m 57s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5434d7b3e40942d7803b618918f39d15 : SUCCESS in 1h 05m 20s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/63b3afa7bcf0448f95dec8591fcd9164 : SUCCESS in 6m 17s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/003e84186b8248e8a64d168b581fc6af : SUCCESS in 5m 05s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/10391b9f109a44a088dbfd9e8c961956 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9396ef9f9e094a90a0430406bb7ef36b : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12928fda6d594995a6f9be0faeb61deb : SUCCESS in 4m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40b9f4fda2ad480ab6d5192090985e08 : SUCCESS in 55m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/081a582cb8cf451dae013a44909a858e : SUCCESS in 51m 07s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/b80ba017232849eca0fb69a59809b0cd : SUCCESS in 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"33c2dc85c272a5c9a8ca0890d38ca0323c33f2d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:53:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8889666dd769fc7a84dd4e9e033d3c27045ec6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:55:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da7eafb88d7d4d74b62b301cb80364d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15beb9708a54b96ab717e428c506d04 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"659737b716494d6f273881a684efa73f44c02822","revisions":{"659737b716494d6f273881a684efa73f44c02822":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:27:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/1003770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/70/1003770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/70/1003770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/70/1003770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/70/1003770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/70/1003770/1"}}},"commit":{"parents":[{"commit":"b207acd9c6776848fa498c5cf7b924514856df95","subject":"config: Add missing docs for CloudRegion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b207acd9c6776848fa498c5cf7b924514856df95"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-12 10:27:55.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 12:46:08.000000000","tz":60},"subject":"config: Separate parsing of .client section","message":"config: Separate parsing of .client section\n\nChange-Id: I4de8a2852782d70710911f46e330a74978590829\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/659737b716494d6f273881a684efa73f44c02822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/659737b716494d6f273881a684efa73f44c02822"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
