)]}'
{"id":"openstack%2Fopenstacksdk~1003771","triplet_id":"openstack%2Fopenstacksdk~master~If749125d306271156d9e2ed45256ea32c05f7af1","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 14:19:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"If749125d306271156d9e2ed45256ea32c05f7af1","subject":"config: Separate parsing of .metrics section","status":"MERGED","created":"2026-09-03 13:27:58.000000000","updated":"2026-09-07 15:56:06.000000000","submitted":"2026-09-07 15:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003771","meta_rev_id":"2df9e14ce66591ecc5199b41734c6a0fa2420849","_number":1003771,"virtual_id_number":1003771,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 15:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-09-04 14:19:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 14:19:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:38:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 14:19:08.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"6ff445473ab30fd2ef2401bf4ecf56cab5778e11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"716019a50bfe9dc652c30d9b1f09c3ea99d7d91e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:38:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0a561527443415fa0a9d08cb8f99a3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90db33d371fd4a4d8b303f3e5f1ce276 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/783db1414ebe4bf9bcce27bd72bbe4d1 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/476f292248424ccc954f1b569a1f0dd1 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/74ef073bfed94284812ecfd3e312e6a7 : SUCCESS in 6m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a099dd1c5d31470094600b1af4de55e0 : SUCCESS in 7m 35s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/0c02087bc057473fa07e7e51b59df802 : SUCCESS in 9m 47s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/357ac4a893ef428593294b9849fd2e98 : SUCCESS in 1h 09m 10s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9d058f9eb50c4a64ad83000dbacbbac7 : SUCCESS in 6m 06s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0c61806e9a3849ea9df4e519792cdb8c : SUCCESS in 5m 30s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/3f9f7d93271e4d90b18c88ec145d8c24 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54b635e1570c48e288aa0c5eeb822ee4 : SUCCESS in 15m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6d4f3c8b3df45c694afac6dc84e9aad : SUCCESS in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/688169f1a7444cd682110dd7660d96b4 : SUCCESS in 1h 03m 05s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c9527827731847fb8364cb7d5dd2d741 : SUCCESS in 1h 34m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/2ce173d03332448fb2879b09bae25830 : FAILURE in 1h 24m 42s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a4c3117df9a84cfd9f3edff642493662 : FAILURE in 43m 01s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/61ef53d347b145fb997624dee4ab4948 : SUCCESS in 37m 25s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/d744504616e148bf9854144673d23718 : SUCCESS in 32m 57s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0d32d141ff8d4b49a71bfc949c8b5463 : SUCCESS in 29m 12s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/33f047680c134a9c9d70ec523b5c0248 : SUCCESS in 33m 10s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f4e7bb35176c413296a97d46b6884000 : SUCCESS in 51m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f9d1ac2c47b74d3d9356e411590ef7de : SUCCESS in 1h 14m 01s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/8bbfdcccaed545fd9c47566bb88aa147 : SUCCESS in 1h 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1faab7577b5f9244f2143b20a77e7ce812771859","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 14:19:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"09e8e519df2fd18cbfcb8dde107e8b93847b0bed","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 14:19:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"752fe3ba439a9f6df5db73c4e8f6babbe160568d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 14:21:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd24068d38e66d3ca84d871f84d52d8960cbf347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:53:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d1e1dbebb14b0488409b2ca8ac86fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef14ec52752a42c5b7d7fd73a7f0fbbe : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f8865b3114a43df8357599238c3fdc2 : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d9c521002ad4dc2aa5ac51873a38e5b : SUCCESS in 5m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c88aa00bdb0948299ebfc6846f792ba1 : SUCCESS in 7m 10s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e713d40b7a7343b89ed4cb02ed5a0e8d : SUCCESS in 57m 44s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/97cbfc2cafa74c4da43cbc702a02d364 : SUCCESS in 2m 32s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7c5654f17d8d4457a16d664cf8e4e4cd : SUCCESS in 5m 12s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f289f76179904de4b0434fa10979e98d : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d11dab89e424bf4a5a56d9a9cc3ef3e : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4196e157add42ee949222fc2e1730c3 : SUCCESS in 5m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dbea5e05187441a86e06702c32692ba : SUCCESS in 1h 01m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/cef7f7f9e70047e59f758755db20a16c : SUCCESS in 1h 21m 45s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/a5cf088836fd44e995a604094b639868 : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cb48bb528187784f5a89b84682d06bc4af4db9f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:53:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2df9e14ce66591ecc5199b41734c6a0fa2420849","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:56:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7b79bd12d64ab18dee321798f0131b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7b46f15c5a40ddb94b7c21fddffef1 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0c8c9ad91541899daaf1cda9700b01 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"549a741d6147eb2e753457708d0b3c5892b85774","revisions":{"549a741d6147eb2e753457708d0b3c5892b85774":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:27:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/1003771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/71/1003771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/71/1003771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/71/1003771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/71/1003771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/71/1003771/1"}}},"commit":{"parents":[{"commit":"659737b716494d6f273881a684efa73f44c02822","subject":"config: Separate parsing of .client section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/659737b716494d6f273881a684efa73f44c02822"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-12 11:17:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 13:27:52.000000000","tz":60},"subject":"config: Separate parsing of .metrics section","message":"config: Separate parsing of .metrics section\n\nChange-Id: If749125d306271156d9e2ed45256ea32c05f7af1\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/549a741d6147eb2e753457708d0b3c5892b85774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/549a741d6147eb2e753457708d0b3c5892b85774"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
