)]}'
{"id":"openstack%2Fopenstacksdk~1003773","triplet_id":"openstack%2Fopenstacksdk~master~I8a143fc5414381b1f10ee85211f854cc127c3922","project":"openstack/openstacksdk","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-03 15:28:20.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I8a143fc5414381b1f10ee85211f854cc127c3922","subject":"config: Separate parsing of .cache section","status":"NEW","created":"2026-09-03 13:27:58.000000000","updated":"2026-09-07 15:44:45.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ce525bc973c0a21df4e6b2effdd330d1c46c32cb","_number":1003773,"virtual_id_number":1003773,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-07 15:44:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea50197b9e92728852d1ca85d51d42e8cf750c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71d636cb4c2adcb83303140c969ac24ecf0ae18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:28:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ac5f389e1734f8d97a147d46acb1717\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2279f7b9f524450bcf0c10181621d2f : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf380eedf20c412d9c12aea033f0687a : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ea2063cdacd4edb91cd2a0124014c1d : SUCCESS in 5m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92a17bd34ff04a318ce37279d12fbf28 : SUCCESS in 5m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a772b8ed32374c68a1b2ef11954def7e : SUCCESS in 4m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c2eea9d19b734d9aab7ce37130a0a945 : SUCCESS in 7m 52s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9a1a5b86c0ef48ed9c12b4e9dfbeb186 : SUCCESS in 1h 10m 26s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/23db40e1aff64d4fb3a02723810e731b : SUCCESS in 4m 24s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/03b6ba45184540bfae01ac30282b1e9f : SUCCESS in 3m 35s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/57d10f590092408097b7bce3a9e68e32 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aaba60b29ff48d79d9ef8cc27b7914b : FAILURE in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3e39f0104da431bbabdb6b94e0f3437 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b04981b90c0749bf8e8219bad9bdb845 : SUCCESS in 1h 07m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8c0a76d6221941c49816aae6a333c350 : SUCCESS in 53m 04s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/380feea6f9344f4aa3030f57cac45739 : FAILURE in 1h 26m 54s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/867b28ac7b08494e8658c39f59dea4eb : FAILURE in 1h 01m 18s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/57540d448b0045b4b64bef526ec4f267 : SUCCESS in 22m 03s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/e93039cac4684c7b9aaff9f8abd61a97 : SUCCESS in 20m 00s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/7784a7870a4a4cc4beb1e9a1208f8873 : SUCCESS in 14m 30s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/7353ca35f8774c4487504958ba378482 : SUCCESS in 38m 02s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/11e1aba47439479b987aef9151ed5a44 : SUCCESS in 45m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ac1abbd888a944ca987fdcdcbe0b4c9a : SUCCESS in 1h 24m 50s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/416a5f82cfc241df99b86622443c69db : SUCCESS in 1h 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db3ca3412d29e994cfbd2f7e4195bd2c412cc24f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 14:22:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ce525bc973c0a21df4e6b2effdd330d1c46c32cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:44:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74224bfa6af14cdfbe0e223c6b5138ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a8ea4cc24ed46378776f882b099a5b5 : SUCCESS in 6m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5572f7d0b2a0499389b88a199aa0db90 : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d165cbb882849d59b7a5dcd7b4f72f5 : SUCCESS in 3m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63437fa2a8de4b709f12b5d495ac78b0 : SUCCESS in 4m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7fba4b0112364148a0fe46578aeadcb7 : SUCCESS in 5m 00s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/03893f26451945558e8baaf1c4453827 : SUCCESS in 7m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a9e85243686f4e26895c4cd2035ddb81 : SUCCESS in 1h 02m 05s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f92bc89f6f1648e0b3192cbd5584bf94 : SUCCESS in 5m 32s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c5be362fd0bf42448392bd229b57a4e2 : SUCCESS in 3m 50s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f5a1cf4372f545278f1e8d2574eef021 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f8b6f641524905a30e07bd134bef25 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0b900afbe114e998985b69f0b6a5683 : SUCCESS in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4405a9d379474b53874c74459c64c166 : SUCCESS in 56m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/240c7703197745769c02106448c95fd4 : POST_FAILURE in 11m 44s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/e361bbb342d244bfbcfc51ac4dc38cad : FAILURE in 59m 26s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ccefd39c94584747bbb58559600a40d6 : FAILURE in 36m 32s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/253b932e924d4858b4f3b9b4ce1ebe16 : SUCCESS in 35m 01s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/af6d516df10049d4a736191152ad2681 : SUCCESS in 33m 35s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/17ee7b18f6484710a9c4b258e5c67c14 : SUCCESS in 27m 12s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/d4a9dbc3c2d3496c85dfeeb5fe2da194 : SUCCESS in 35m 39s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7d5cd1ef0f4a4fa78ddea69b09ea7de6 : SUCCESS in 41m 44s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/eb16e422be084579be92b4ec88ef4f31 : SUCCESS in 1h 14m 42s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/dbebb708a17b442783f1e19659fd030d : SUCCESS in 1h 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d30925a168ec74479c1ce9b7afdcbbaf45fab07","revisions":{"fb6a764fe9511fb0c5a5bdb867bc7db72cc184fa":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:27:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/1003773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/73/1003773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/1"}}},"commit":{"parents":[{"commit":"22ffd4e1d25c9d9b66615dd61892450277ca36ea","subject":"docs: Update metrics guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/22ffd4e1d25c9d9b66615dd61892450277ca36ea"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-20 14:51:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 13:27:52.000000000","tz":60},"subject":"config: Separate parsing of .cache section","message":"config: Separate parsing of .cache section\n\nExtract the inline parsing of the \u0027.cache\u0027 section of clouds.yaml into a\ndedicated _parse_section__cache helper returning a CacheConfig typed\ndict, mirroring the existing _parse_section__client and\n_parse_section__metrics_* helpers. No functional change.\n\nChange-Id: I8a143fc5414381b1f10ee85211f854cc127c3922\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fb6a764fe9511fb0c5a5bdb867bc7db72cc184fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/fb6a764fe9511fb0c5a5bdb867bc7db72cc184fa"}]},"branch":"refs/heads/master"},"5d30925a168ec74479c1ce9b7afdcbbaf45fab07":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-07 14:22:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/1003773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/73/1003773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/73/1003773/2"}}},"commit":{"parents":[{"commit":"2e5f603ec0ca1a26e2eae978219df127773729e7","subject":"docs: Update metrics guide","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/2e5f603ec0ca1a26e2eae978219df127773729e7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-20 14:51:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 14:22:29.000000000","tz":60},"subject":"config: Separate parsing of .cache section","message":"config: Separate parsing of .cache section\n\nExtract the inline parsing of the \u0027.cache\u0027 section of clouds.yaml into a\ndedicated _parse_section__cache helper returning a CacheConfig typed\ndict, mirroring the existing _parse_section__client and\n_parse_section__metrics_* helpers. No functional change.\n\nChange-Id: I8a143fc5414381b1f10ee85211f854cc127c3922\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5d30925a168ec74479c1ce9b7afdcbbaf45fab07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/5d30925a168ec74479c1ce9b7afdcbbaf45fab07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
