)]}'
{"id":"openstack%2Fopenstacksdk~1003921","triplet_id":"openstack%2Fopenstacksdk~master~I7f8da39642f8654da39062e57f2c8f3f92dd03af","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 11:14:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-07 13:00:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7f8da39642f8654da39062e57f2c8f3f92dd03af","subject":"fix: Close underlying keystoneauth connection on exit","status":"MERGED","created":"2026-09-04 07:57:39.000000000","updated":"2026-09-07 13:01:50.000000000","submitted":"2026-09-07 13:00:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1003921","meta_rev_id":"ec8944444ce13ddb69f3660dfa6231e97116cf7c","_number":1003921,"virtual_id_number":1003921,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 13:00:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 11:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 11:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 09:17:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 11:20:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"461873e882eb8a333718e5472a33dc5fd3d5b486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 07:57:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0270115ce03c50d443fc54690ee35ea183925f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 09:17:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/417e61967952496fb5837739875602bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54f0e7c819ec4c55956943e9f90bb083 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e90f0643814541758530146e8bc7494f : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38e2fe247d7d43c48df74dffd215e0d2 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe75453e444c440eb680aa356acb83e5 : SUCCESS in 4m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e96dff5a3134fcf8c5dac487542f585 : SUCCESS in 4m 13s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/7e42e5dfb5d543c8bf77dabc376f60c8 : SUCCESS in 6m 56s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7b6c0a062c634eb5965d1eac348c084a : SUCCESS in 1h 03m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e0a47aaa87a24d62b96195e7df1e540b : SUCCESS in 4m 30s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/584e801aaa2c46f49145955f6a28a03f : SUCCESS in 4m 35s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/bebbbb7f45634e6cb03c7cd02322e3b6 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8234fdfb484d66bd470b6c877ce254 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bbfc974e6444b699a94c7b4b03ac297 : SUCCESS in 6m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcf3e80ea1fc4f22afff4e546ac32e01 : SUCCESS in 29m 11s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6fd38468393a4c5989262c37420be61a : SUCCESS in 1h 14m 11s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/319aafcc12cb47979a291d5a55eb003d : FAILURE in 53m 35s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2b4485206aaa478cb19785d219b2abbe : FAILURE in 55m 55s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/7ac123c63b704e99b56bb7ff04ca9c82 : SUCCESS in 20m 05s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/72b160ef12374531b616f12af0f53a75 : SUCCESS in 19m 02s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/50530089a5c640e49a084d427b05bafa : SUCCESS in 27m 55s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6a4148cda3474c2e8157ecbbd3760686 : SUCCESS in 13m 36s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f5cb95c062c147d2a89526bc214688b3 : SUCCESS in 18m 44s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/eb47b278a51d4990b70ea731312da612 : SUCCESS in 1h 10m 59s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2b0e36ba4b8f4a9d9eef5f2556a0a1f3 : SUCCESS in 1h 15m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b216978968a4dfb1546158d301768ef4435dbbde","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-04 11:20:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"52c54588b4837b18624f3e84180675837dcd8dff","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-04 12:53:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"773818d4a858ade163ab59ed13c6168f45ce8a1f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 13:26:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"536da64736afc1033291da2fad1e4dd2510e6582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 17:52:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"803ce2cde7f83455dc196e10e97f27110df60a75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 19:21:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4b03fea570748f6a867e8b9cbbcf527\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e9ebd5684ca4158a8e1044545488233 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/652ef3e561e44782bbbdc95613a8712d : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3dba71a120a343fdbbc53182095f3c8c : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b094695ec46c4c2487603b2e16b60899 : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1cdcacaaef9d41dea6f2116d6147cbe6 : SUCCESS in 5m 07s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c3aca64984564d6491020d914f179e47 : SUCCESS in 7m 08s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6064ddd0a4464ac6b95d0bf49f6ed520 : SUCCESS in 29m 43s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8b4d2df3466744bfbce074c6613186e5 : SUCCESS in 2m 29s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/492e82841f57450283d37a8e455b0d24 : SUCCESS in 2m 42s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/36c35ef5d0ea4720bcf13c8c05b30420 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be92319895a648f98f08991199a6d649 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a5ea1fdc71c4704a1c52aa0a22cb26d : SUCCESS in 3m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d97e7e5859f748138b2087a90f75a8ab : SUCCESS in 52m 40s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/e5b80b0628344c9fafdeff81fb2fa8dc : SUCCESS in 1h 04m 49s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/d30138d0ca984914a669876f35881c91 : FAILURE in 59m 53s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/c98f313b0b3449d6b7700808e8085eaf : FAILURE in 33m 37s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/452b209d3d354115872d5a76edb1a12b : SUCCESS in 20m 09s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/1bfffd4586d1498a88fea13d4b847270 : SUCCESS in 24m 00s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/6c9f0ef684c743d3bf4314b03bb1521b : SUCCESS in 28m 12s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6b3587d1bc8d444b804f01a077be5e20 : SUCCESS in 31m 03s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/dcf34bdac73e4e028065a84fd66a8cf2 : SUCCESS in 44m 15s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8e2fa2221c83493a9c2d1f7248cd29aa : SUCCESS in 1h 12m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/caa5b5886cf24f2bb4bba5fb775bf92d : SUCCESS in 1h 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ef03066484914b32d8ee4cf347198179a96abb1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 11:14:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"952620f999a1a01bd8975a110d337215db7af365","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:15:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c5abbc01ebd93c11d126828f472e67bb57497fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:00:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb8e73bc9c04a8fbd4d920860952745\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d8cc8471e2a4041872702ffd937aafc : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/571e90f6a7434544b491a28693aed13f : SUCCESS in 6m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7d49b30189c24e7f9969546e491510ee : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d165d6413da347fc9c622cea08bed658 : SUCCESS in 6m 46s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3104c9553b7842cbbc479ad6a83e9036 : SUCCESS in 1h 00m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1f72c786b40b4b28adfbd83e51069ea6 : SUCCESS in 4m 03s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a47047bfb7ab4150a29f424d7749f59f : SUCCESS in 4m 58s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/03c0341afd2844fa9cfe1d6a6200f1e5 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2758801a0f84396b0ff3d5b0d992802 : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2279f0e06344dc5a1fcc908e625c8af : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1af7ba3ae4a4444592adec24217e30ad : SUCCESS in 1h 02m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/688f03c0a82c4568b6d651d90dc00dce : SUCCESS in 1h 39m 48s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3ca9ddf4b31c4ed080f5e2b2a2eaffdd : SUCCESS in 43m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8b9ab952bc2d6d99869a1fc4725c1a1f48efb106","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:00:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ec8944444ce13ddb69f3660dfa6231e97116cf7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 13:01:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93db0a59d6204768806a4f92b64dbb2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5261abce2745228d42c4d08a077cb7 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfba553ffccb493dbde02614c1bd042c : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c49994ac1045f810b347106ef5ea38fb4cd950f","revisions":{"612d09fd8371f3913bfce2e19e6b3e2931b77419":{"kind":"REWORK","_number":1,"created":"2026-09-04 07:57:39.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/21/1003921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/21/1003921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/1"}}},"commit":{"parents":[{"commit":"9366bde3a247ca044614a5c78f989f12b726ab18","subject":"Merge \"config: Add missing docs for CloudRegion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9366bde3a247ca044614a5c78f989f12b726ab18"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2026-09-04 07:53:56.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2026-09-04 07:57:37.000000000","tz":120},"subject":"fix: Close underlying keystoneauth connection on exit","message":"fix: Close underlying keystoneauth connection on exit\n\nWhen a long-running process explicitly closes openstacksdk session the\nOS socket remains open, since keystone1 session is not closed which\nholds the requests session, that in turn holds the socket. This is not\nvisible by scripts, but is very visible in Octavia that creates and\ncloses session in the user context. Explicitly try to terminate the\nkeystoneauth1 session (in a separate change dedicated \u0027close\u0027 method\nis added there as well).\n\nPartial-bug: 2166351\nChange-Id: I7f8da39642f8654da39062e57f2c8f3f92dd03af\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/612d09fd8371f3913bfce2e19e6b3e2931b77419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/612d09fd8371f3913bfce2e19e6b3e2931b77419"}]},"branch":"refs/heads/master"},"0c49994ac1045f810b347106ef5ea38fb4cd950f":{"kind":"REWORK","_number":2,"created":"2026-09-04 17:52:02.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/21/1003921/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/21/1003921/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/21/1003921/2"}}},"commit":{"parents":[{"commit":"9366bde3a247ca044614a5c78f989f12b726ab18","subject":"Merge \"config: Add missing docs for CloudRegion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9366bde3a247ca044614a5c78f989f12b726ab18"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2026-09-04 07:53:56.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2026-09-04 17:44:59.000000000","tz":120},"subject":"fix: Close underlying keystoneauth connection on exit","message":"fix: Close underlying keystoneauth connection on exit\n\nWhen a long-running process explicitly closes openstacksdk session the\nOS socket remains open, since keystone1 session is not closed which\nholds the requests session, that in turn holds the socket. This is not\nvisible by scripts, but is very visible in Octavia that creates and\ncloses session in the user context. Explicitly try to terminate the\nkeystoneauth1 session (in a separate change dedicated \u0027close\u0027 method\nis added there as well).\n\nPartial-bug: 2166351\nChange-Id: I7f8da39642f8654da39062e57f2c8f3f92dd03af\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0c49994ac1045f810b347106ef5ea38fb4cd950f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/0c49994ac1045f810b347106ef5ea38fb4cd950f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
