)]}'
{"id":"openstack%2Fopenstacksdk~1003997","triplet_id":"openstack%2Fopenstacksdk~master~I27c93d4f20760c9ff0614dae427dcd8ba3d457e1","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"39399":{"account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"last_update":"2026-09-07 23:47:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I27c93d4f20760c9ff0614dae427dcd8ba3d457e1","subject":"identity: Remove TagMixin from Domain","status":"MERGED","created":"2026-09-04 15:41:31.000000000","updated":"2026-09-07 23:49:08.000000000","submitted":"2026-09-07 23:47:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003997","meta_rev_id":"3f83970636642e0f9191b9f3610ae26836b62d4c","_number":1003997,"virtual_id_number":1003997,"owner":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 23:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-04 20:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-07 22:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-07 22:14:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-04 17:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-04 20:15:55.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-09-07 22:14:41.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"24a6d058dbcfa4ead4fbb3b8d9998287036af59e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-04 15:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17b68fced6de7d27c3f9bd86fd55bdfabb656edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 17:09:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d5b4d986ec84b2f961d33721a7f26c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ca4446a0fa44b9a8a141eac3163d71 : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26294b4a4eaf4c1c8585b958017a252b : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e0c30f02e78483e826e8f8964f9caca : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9d288249b604e5eb2393826a0396622 : SUCCESS in 5m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/240504858aa64fdda486d65ad8c82c57 : SUCCESS in 6m 42s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/de9fd119440240c7b4fbe0ef66c420a9 : SUCCESS in 6m 13s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cfa5f2ac36dc44c7803b28d2cf04375b : SUCCESS in 27m 39s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cc6ae1481dff46c091fb2311ef68bed6 : SUCCESS in 3m 13s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/927e20e95efb4c2baf29e08800cabbf7 : SUCCESS in 6m 56s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6d309e0f8b014e2894fb02fc8f9a24ac : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4cf07434984a0abd65f64d77208422 : SUCCESS in 11m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a571375984ca4545a7f4279b2d3d6f78 : SUCCESS in 4m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/405ee875fb1b49058b853df641c55462 : SUCCESS in 54m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f2fcc74b32e64dff9fcb5966f83c40ce : SUCCESS in 1h 05m 32s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/1aed8bee9e2d4d51ab765863f986c1fb : FAILURE in 1h 15m 59s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/68a1c62c81714de3ae04abeecba2548b : FAILURE in 58m 16s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/65ecc34272e04099a799270619ae1fb2 : SUCCESS in 25m 26s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/bf1fbe0a553240dba011cf9e19c56dc6 : SUCCESS in 29m 59s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/8d2c3b805676414bbb8215e73924b29a : SUCCESS in 29m 38s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/70f24ffa2c804716a88ba49eb62782c8 : SUCCESS in 30m 50s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/0d88f464598548999c13ece9b086e882 : SUCCESS in 40m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ed494348376c4f6d88485cfc67244e95 : SUCCESS in 38m 35s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/6dbdc6f2fb7e446a988778018b65b691 : SUCCESS in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a41ee1198e47afb1250da929832373e41b09d7e2","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-04 20:15:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aada89008afd95f63d8a96ddf6a5b37e5720bc09","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 22:14:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48fa8313d046fff15219ef4bc95a5dc5faf80dac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 22:15:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e69c1be274bf6837063cfbc1df6995dc0403ab03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:47:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0da5c386424028814b869d2b484a90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6494a95d2ea46a6b89d7e72bfe354b4 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/496c45da25384520916e1fe4327342cb : SUCCESS in 3m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/51c51116b73a489d98bdab612b952abf : SUCCESS in 2m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fe62c72e3bf48fa9779ffbce11d9b45 : SUCCESS in 7m 29s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a489ca1e086c441090c5fe162978eccf : SUCCESS in 32m 36s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ad9211719cba4c50bb9f8025d0fd8b01 : SUCCESS in 5m 18s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e60bd70dd53b471a91be01ef041a24dd : SUCCESS in 4m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/0f7e25bc06ba4c89935298e088ba2364 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a215ffb67a13464e8387f75f4c43e1d4 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce776ba69f0843f992a7d186e52820f0 : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73889e293e5444edae500670411c8116 : SUCCESS in 40m 52s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/422c202fdc6e473b8a848357f1a54ce2 : SUCCESS in 1h 20m 01s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/fc152f7938d2404ba350d9f8ea087a37 : SUCCESS in 38m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1cc79a1e5a033cacd76390c7978f45656f92fe6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:47:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f83970636642e0f9191b9f3610ae26836b62d4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:49:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce77ec2459de405c83b5f22ece6bbbd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a11543b61f4d5bbac66b1a77b52cda : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39650899b48f4b7cbfb633d2d1aaa8a2 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e251f957d2011e37dc3a6030df55ba1772051c96","revisions":{"e251f957d2011e37dc3a6030df55ba1772051c96":{"kind":"REWORK","_number":1,"created":"2026-09-04 15:41:31.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/97/1003997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/97/1003997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/97/1003997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/97/1003997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/97/1003997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/97/1003997/1"}}},"commit":{"parents":[{"commit":"9366bde3a247ca044614a5c78f989f12b726ab18","subject":"Merge \"config: Add missing docs for CloudRegion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/9366bde3a247ca044614a5c78f989f12b726ab18"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-04 15:34:44.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-04 15:40:38.000000000","tz":420},"subject":"identity: Remove TagMixin from Domain","message":"identity: Remove TagMixin from Domain\n\nThe Identity API does not support tags on domains.\n\nCloses-Bug: #2166512\nChange-Id: I27c93d4f20760c9ff0614dae427dcd8ba3d457e1\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e251f957d2011e37dc3a6030df55ba1772051c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/e251f957d2011e37dc3a6030df55ba1772051c96"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
