)]}'
{"id":"openstack%2Fopenstacksdk~1004357","triplet_id":"openstack%2Fopenstacksdk~master~I8fd5372db61dd5c9e110717440575e8886915684","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 10:02:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-07 11:29:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8fd5372db61dd5c9e110717440575e8886915684","subject":"Fix typos","status":"MERGED","created":"2026-09-06 22:43:01.000000000","updated":"2026-09-07 11:31:00.000000000","submitted":"2026-09-07 11:29:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004357","meta_rev_id":"d7b24c4583cd9a18199da197735d95991f5d32c3","_number":1004357,"virtual_id_number":1004357,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 11:29:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 10:02:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 10:02:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 22:44:48.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-07 02:26:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f71d97fae39caf1e2bec454278cd452664ffa36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:43:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"975cd851dc5bbe8e7a5fb30bd135e9dd7a82346a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:44:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9ddb793430eda14c1c34a1f1c3158cf120617424","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:44:48.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"ac3aef140a7cc947db28f5753b3c9e7cc8e3515d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:08:06.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2073a8a6a8c82932f3232d04157297ee306b8c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:08:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"03731fb355e9b25b3966ee5a18eb9618a505adff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:54:10.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b84e0e3345203c10e2e56b66532f56f11b2772b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 02:26:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a9c687097d641a2b534a2bb3e64e8f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a3cc3b145542b990acccb9b0e5fb43 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba9e61f42874480e95f18632c005f516 : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85116018195741e684610aebffb03c84 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27f2c150801247d0add67d0f25926616 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f54d59c8025e4294ac2516d226f0ea74 : SUCCESS in 6m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3c2dad0d767346279428da150a992f87 : SUCCESS in 8m 46s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ecef8ba703674c9eb17bccf4e7b0eaea : SUCCESS in 31m 58s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3ef061ef51cf41a2b630ddce41b4f8c1 : SUCCESS in 4m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/79db09c5700e4d8287472ecc5a17f6a2 : SUCCESS in 3m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/d2921d211ea44ef28108277366552b54 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84cabec0359f41bda0b7c73728b20575 : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11f73a7e23854146876f3411ba2debea : SUCCESS in 5m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69c34766f5e94811a5d5075236992b94 : SUCCESS in 32m 41s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/f550e006732344d3bc394143146d10b7 : SUCCESS in 56m 14s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/12b302106e114d94a42366d71bdcd8ce : FAILURE in 1h 26m 37s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2f928df8343c4fc690ccddbbe88b9b0d : FAILURE in 26m 27s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/557a703c20064340a1e18d3b631843f3 : SUCCESS in 27m 33s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/36364021fe0646da9c8a842cbeb8471b : SUCCESS in 30m 41s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/8908d999886849b983c21956bedffe85 : SUCCESS in 15m 26s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/6888bc32d0a74d7ca72edc617927fa53 : SUCCESS in 29m 09s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/3346f59bda3740c29cf840deece0c9f5 : SUCCESS in 40m 18s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e88d2960625743498516943a77a5ea6c : SUCCESS in 1h 10m 09s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/c2d47c029617439fb6e81bb300373d2e : SUCCESS in 1h 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e696eb535614aff0ff3b0400f97760dea84ce601","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 10:02:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"3c70347a08218b1db4fa20b7b2dfa946732dbf75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 10:03:54.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7c605dcfbc2af690ac276e3b6eea7439226ea654","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:29:42.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918a333f0967462e96e4b16b0ab04ad0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a86c3e6a834c27a771d97d28ca6fce : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3d893c32f1946e0b6a37bd19cbcafb8 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/789f77ba8fad408f997043d20878920c : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4f99568122a345ee97495de635dcbf7e : SUCCESS in 6m 21s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/10ea33c7da294c0f95757a9d3609bdc6 : SUCCESS in 59m 20s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/954f636905e74a7fb1ae8eef7ad50e06 : SUCCESS in 3m 52s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b68d4b27f0a045da82a46ce11674b535 : SUCCESS in 5m 10s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/565ad5ac37664704806f3e1164043837 : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b310108563264e4ab450ac827048a246 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbb5d19f3dec430db1beaeb60c6bc2ae : SUCCESS in 4m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d52ff558005d4b89829658903b89aab6 : SUCCESS in 55m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/53dfeae35b3547a4b2a0758d4e8d3284 : SUCCESS in 1h 22m 04s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/63c220e75b1544aba316c14054ad55ee : SUCCESS in 40m 45s","accounts_in_message":[],"_revision_number":5},{"id":"28b59790057b723ee4e2208ac1c4a175511980e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:29:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d7b24c4583cd9a18199da197735d95991f5d32c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:31:00.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8ae01faca664e1cbe5744adb65293d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2895cdf1b24156b73b35e590efef59 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f9fe33bcc8bb4fc618a31c49d2bf2baa70cecdc8","revisions":{"042e2fd1a229b48bff2a18699e4e0d9895be1cfe":{"kind":"REWORK","_number":1,"created":"2026-09-06 22:43:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/1004357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/57/1004357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/1"}}},"commit":{"parents":[{"commit":"f1ab1c9b4482fc159c55b2f2378572cce6be3595","subject":"Merge \"stats: Fix prometheus histogram units\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1ab1c9b4482fc159c55b2f2378572cce6be3595"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:43:01.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I8fd5372db61dd5c9e110717440575e8886915684\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/042e2fd1a229b48bff2a18699e4e0d9895be1cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/042e2fd1a229b48bff2a18699e4e0d9895be1cfe"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"7a3cfb09da1a5fb4a5d8f1c27f5297f7be64769e":{"kind":"REWORK","_number":2,"created":"2026-09-06 22:44:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/1004357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/57/1004357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/2"}}},"commit":{"parents":[{"commit":"f1ab1c9b4482fc159c55b2f2378572cce6be3595","subject":"Merge \"stats: Fix prometheus histogram units\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1ab1c9b4482fc159c55b2f2378572cce6be3595"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:44:43.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I8fd5372db61dd5c9e110717440575e8886915684\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7a3cfb09da1a5fb4a5d8f1c27f5297f7be64769e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7a3cfb09da1a5fb4a5d8f1c27f5297f7be64769e"}]},"branch":"refs/heads/master"},"b5237befebcd7c16e98267ef69ba3f1b2ff0c7cd":{"kind":"REWORK","_number":3,"created":"2026-09-07 00:08:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/1004357/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/57/1004357/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/3"}}},"commit":{"parents":[{"commit":"f1ab1c9b4482fc159c55b2f2378572cce6be3595","subject":"Merge \"stats: Fix prometheus histogram units\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1ab1c9b4482fc159c55b2f2378572cce6be3595"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:08:05.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I8fd5372db61dd5c9e110717440575e8886915684\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b5237befebcd7c16e98267ef69ba3f1b2ff0c7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b5237befebcd7c16e98267ef69ba3f1b2ff0c7cd"}]},"branch":"refs/heads/master"},"c4adce890df7f1fd47cd51e0d0eab4a0cbf260ed":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-07 00:08:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/1004357/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/57/1004357/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/4"}}},"commit":{"parents":[{"commit":"f1ab1c9b4482fc159c55b2f2378572cce6be3595","subject":"Merge \"stats: Fix prometheus histogram units\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1ab1c9b4482fc159c55b2f2378572cce6be3595"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:08:38.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I8fd5372db61dd5c9e110717440575e8886915684\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c4adce890df7f1fd47cd51e0d0eab4a0cbf260ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c4adce890df7f1fd47cd51e0d0eab4a0cbf260ed"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f9fe33bcc8bb4fc618a31c49d2bf2baa70cecdc8":{"kind":"REWORK","_number":5,"created":"2026-09-07 00:54:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/57/1004357/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/57/1004357/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/57/1004357/5"}}},"commit":{"parents":[{"commit":"f1ab1c9b4482fc159c55b2f2378572cce6be3595","subject":"Merge \"stats: Fix prometheus histogram units\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1ab1c9b4482fc159c55b2f2378572cce6be3595"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:43:01.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:54:09.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I8fd5372db61dd5c9e110717440575e8886915684\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f9fe33bcc8bb4fc618a31c49d2bf2baa70cecdc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f9fe33bcc8bb4fc618a31c49d2bf2baa70cecdc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
