)]}'
{"id":"openstack%2Fopenstacksdk~1004427","triplet_id":"openstack%2Fopenstacksdk~master~I10a00711df937711e8ad5bf3b485fc6656d28bee","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I10a00711df937711e8ad5bf3b485fc6656d28bee","subject":"typing: Resolve TODO","status":"MERGED","created":"2026-09-07 14:17:18.000000000","updated":"2026-09-07 23:26:57.000000000","submitted":"2026-09-07 23:25:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004427","meta_rev_id":"7b68d38f199bfb5aae0a9bec99a981ef5f9e1f70","_number":1004427,"virtual_id_number":1004427,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 23:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 21:56:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 21:56:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 15:50:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac69c5c4f0e8470046d0cf65eaaadb1f77ccbd09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 14:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3284d929f39b03927b0f4e52fc7f771446c191e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:50:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af74148f2f274692a74fd05ed0372261\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64d2660dcd9241e39d012d38b97d6bb6 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df7fb362c99c43af9abf2545b747479a : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6db630690e3b4e08825f25db347479f1 : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/688e4333a2bc48d3833f510a2207c54a : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c262238d065a4b0fb783584344625ddf : SUCCESS in 7m 10s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/1eff3d7d304d4f9f9124cd913d179883 : SUCCESS in 7m 05s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/40bf9ed4f560412196bbe24b4153d8ad : SUCCESS in 1h 04m 22s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/59e639552c2444e9b83a90cd086f02cb : SUCCESS in 3m 48s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5365cbab094e444d92e74c27244a9e58 : SUCCESS in 4m 32s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/aac8e2a6304f46728f2fdde7a65127a6 : SUCCESS in 10m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71d51d2e2394e3e905d6e14d9fab121 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54da25731ab148fb8cb30ae9aee50290 : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c863c38fca14be29ff06118669419b1 : SUCCESS in 34m 56s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6ea9256975b046b78af0d7e12d0ae505 : SUCCESS in 51m 19s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/0d7ea7dd244b46059168275b2558ab76 : FAILURE in 1h 26m 40s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/684fb22d2e594e8ca36db43e98fe35a1 : FAILURE in 30m 14s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/495947b1316a4c62b6e3984903ae4304 : SUCCESS in 23m 18s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/c4c3e99f5a6a418f834f013270b52058 : SUCCESS in 27m 48s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/6e5d216f61f143b390f3aae2a4a4df69 : SUCCESS in 13m 13s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/054eb4525ef64794adeb5c478d6ee9ef : SUCCESS in 24m 20s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/22662dc11a534ede8ed1c07db2059a5a : SUCCESS in 18m 55s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/aff91084b9b14665ae52044a03fd765f : SUCCESS in 56m 20s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/92873e130e5149708c140e01f0bab945 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b64338a17cbdb007535641fd2fbf3e650e7feeda","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 21:56:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7beba39bbf8affbb870d3bd5e192f4c3024537a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:57:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce226831a92381acb9c4dec27bfd67b82bf117a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9490286a9097455e9b361d1b77587b3e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06ddd37019b45c3b0c8c53a005362dc : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc7fb5c7478f485ca273eb4c59465f46 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/399d68d428c348d390a8023dec2718d1 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/95678b9be28248189f894dfe70e470da : SUCCESS in 6m 34s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/186ee6388bbe44bab12a1e2960ee42fe : SUCCESS in 46m 50s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4ef9befa07114d62ace0d32823d0d426 : SUCCESS in 3m 40s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/592075196a8943fdbb8ff70fd71ca5fe : SUCCESS in 4m 48s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9ee23ae4335441009d52c0fd02992662 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dda66b9343242cf869c09501c9ce49c : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6b73475d71e44f58580c3a791bcd7a8 : SUCCESS in 5m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9f1dc05b4e54c278d8933cb245460bb : SUCCESS in 34m 04s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/fd0173971f6f422f8fe1b0d86f0d63e9 : SUCCESS in 1h 23m 28s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/667c66d88aec46d0824f0741f60989f1 : SUCCESS in 40m 13s","accounts_in_message":[],"_revision_number":1},{"id":"f631a88be05a6e4af8f9a825b66b40e0fa56baa6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:25:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7b68d38f199bfb5aae0a9bec99a981ef5f9e1f70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:26:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343918b4a7744339b94f799931ef70d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897d5a3dd80248518c5002f7c0140db0 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c50568d2b781c22b4bfdb2ee4b0f70181bfdecc","revisions":{"1c50568d2b781c22b4bfdb2ee4b0f70181bfdecc":{"kind":"REWORK","_number":1,"created":"2026-09-07 14:17:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/1004427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/27/1004427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/27/1004427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/27/1004427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/27/1004427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/27/1004427/1"}}},"commit":{"parents":[{"commit":"f1ab1c9b4482fc159c55b2f2378572cce6be3595","subject":"Merge \"stats: Fix prometheus histogram units\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f1ab1c9b4482fc159c55b2f2378572cce6be3595"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-04 11:55:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 14:16:42.000000000","tz":60},"subject":"typing: Resolve TODO","message":"typing: Resolve TODO\n\nThe ksa fix [1] merged ages ago and our minimum has been bumped to a\nsuitable version.\n\n[1] https://review.opendev.org/c/openstack/keystoneauth/+/951183\n\nChange-Id: I10a00711df937711e8ad5bf3b485fc6656d28bee\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1c50568d2b781c22b4bfdb2ee4b0f70181bfdecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1c50568d2b781c22b4bfdb2ee4b0f70181bfdecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
