)]}'
{"id":"openstack%2Fopenstacksdk~1004428","triplet_id":"openstack%2Fopenstacksdk~master~Ie6a7f2ba3f756da5295fb1aa4458d11aea392ef6","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"Ie6a7f2ba3f756da5295fb1aa4458d11aea392ef6","subject":"typing: Simplify QueryParamters type handling","status":"MERGED","created":"2026-09-07 14:17:18.000000000","updated":"2026-09-07 23:28:06.000000000","submitted":"2026-09-07 23:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004428","meta_rev_id":"0882f8769b63944f658da96207061382202d48fa","_number":1004428,"virtual_id_number":1004428,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 23:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 21:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 21:56:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 15:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adcc60de5a3843df969f28bcc21537a93b4cae1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 14:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7dd51b9050c6ed1df1a5eb465420496f082eaafa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:49:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f231766e1e44e9977a12806976c3b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2acd5a77e0dc4878b2032037d6b4976e : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e97cedfc7734f8987ffbfd580a170ef : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f710857071540d185136ec13196ee08 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ec184371631d4489853a6e75036d0281 : SUCCESS in 4m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5b8ad1fce0ad4f4bb9d30ee663372a38 : SUCCESS in 7m 24s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5bbe951e86fa4cecb7453bc16e0ddbf9 : FAILURE in 7m 44s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9ff3c2ae4bb54b409a9e15fd89439fc0 : SUCCESS in 1h 03m 04s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a1762f4003c4421487a9408745e31842 : SUCCESS in 4m 37s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4e69e7ac03f14e8688829deea26ff704 : SUCCESS in 4m 37s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/882b197b9a79464c925b867e90de5e69 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4841fe5936114b58a65aadd6d672298b : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a408baac9b4a4691be49cb3a945098df : SUCCESS in 4m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2cda0b1c02a4b9c8631152c2ac3f007 : SUCCESS in 1h 02m 42s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b103d07e999948ec90c8572994f542b3 : SUCCESS in 1h 21m 28s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f954dbb06cca4961ae4cc572e7ab19a6 : FAILURE in 1h 25m 20s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/7396d5c00aa4413b8f4054c28d2f0c04 : FAILURE in 33m 00s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/63990cb8510a451fa8d4c2691ae7bb4f : SUCCESS in 37m 30s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/c6c90492821a48039ba0ab7cb589cd81 : SUCCESS in 16m 47s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/d9ed6f4ac83045e18f5db3fa260f48d3 : SUCCESS in 28m 10s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/aff1ee94da81412189762588e880b2f0 : SUCCESS in 31m 15s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7b91d0bf7b1541c3905c5c2e6870679c : SUCCESS in 38m 45s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/58adc3829b2a4f99b996ff13b7017349 : SUCCESS in 1h 22m 40s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/eeb44807ad0846dd84e7b3a8bbbc6600 : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45723a566332608d72eecf5fd943d08ee5f12c42","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 21:56:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5bd4888e59e507e2c12e24a2f2eab5323bebdb83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea103e25b1ccbd8a4b20029cf000b129b2ed458c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:25:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb01b99214047a29a66f270be419c86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6532e9e1808a4f01b0597d2e702f5221 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5aa084a2be90460ea8fe77e274597095 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66ef2534a623443591a33fbe566f597d : SUCCESS in 4m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b22f983a0a3f4aaaba345795b97ae01f : SUCCESS in 5m 57s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4728829e41354427b50e2ad18c8794a2 : SUCCESS in 42m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/41b723a9798d4f74b7e2089adad925da : SUCCESS in 2m 49s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/65beeb3ba9c34221a2cd3ac06cdc531f : SUCCESS in 3m 23s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/26cc9232efbc479183d0b478d0f5c920 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42499d0fa5cd4ac28dc3e85db5546f1f : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2136971f180b402cab01d5df2b28a841 : SUCCESS in 3m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/984fee2cb14146459806a61f19127ebc : SUCCESS in 59m 10s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6e90b06675644fd4809d21e300478f15 : SUCCESS in 1h 20m 16s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7184631de27b4b59ae10c269e98fdc0a : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c97cafed7427f2a73bde41143242c75efe60fbc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:25:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0882f8769b63944f658da96207061382202d48fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 23:28:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d492bbf14c52448b938bdb3e34cc81cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7510a24b01b43c1affc56d90e41ca68 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f6fb581d3d46185eaaa973478f45146387b1dc31","revisions":{"f6fb581d3d46185eaaa973478f45146387b1dc31":{"kind":"REWORK","_number":1,"created":"2026-09-07 14:17:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/28/1004428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/28/1004428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/28/1004428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/28/1004428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/28/1004428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/28/1004428/1"}}},"commit":{"parents":[{"commit":"1c50568d2b781c22b4bfdb2ee4b0f70181bfdecc","subject":"typing: Resolve TODO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/1c50568d2b781c22b4bfdb2ee4b0f70181bfdecc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-04 12:46:21.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 14:16:43.000000000","tz":60},"subject":"typing: Simplify QueryParamters type handling","message":"typing: Simplify QueryParamters type handling\n\nOur tests were testing callables with a type callabe that accepts a\nsingle argument, but in reality none of the code actually used this\nfunctionality. Remove the test and simplify the types in the process.\n\nChange-Id: Ie6a7f2ba3f756da5295fb1aa4458d11aea392ef6\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f6fb581d3d46185eaaa973478f45146387b1dc31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f6fb581d3d46185eaaa973478f45146387b1dc31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
