)]}'
{"id":"openstack%2Fopenstacksdk~1004429","triplet_id":"openstack%2Fopenstacksdk~master~I0933e7ea5834d4390c15b7aa9dd4bc9698ada127","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I0933e7ea5834d4390c15b7aa9dd4bc9698ada127","subject":"typing: Simplify QueryParamters type handling further","status":"MERGED","created":"2026-09-07 14:17:18.000000000","updated":"2026-09-14 22:29:25.000000000","submitted":"2026-09-14 22:27:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004429","meta_rev_id":"ebab9445d072b7f87bbfc302390f12b661eb5375","_number":1004429,"virtual_id_number":1004429,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 22:27:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 21:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 21:09:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 15:51:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4455d4c814a7f2d63bbeff003f74391937e94217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 14:17:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdc72e8ac781ee150d6dab876b03ad015617f523","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 15:51:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01097034e2e14c39a91ab48df72712d4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6b66b9609f40aa9813448bf0a84115 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a600a46933dd49c28cf2eb59701ef03d : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85c827a70c7e4bd4bb2c847214f82c68 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71e3540cff534290b2a2e1a4daa02aa1 : SUCCESS in 4m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6c7dbabd52a4e8198127875e623b9cf : SUCCESS in 6m 39s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/c0cad6fb1209402d85b3479082a20343 : SUCCESS in 7m 18s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/12403ec679004eb0afac4d683e22b1d5 : SUCCESS in 36m 34s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6d734331515b4bd0b27fa914f0305f06 : SUCCESS in 4m 36s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e1efef3619044c9bb270c5f11ba2b62c : SUCCESS in 4m 08s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/06dda190680b4fb688c547e522e5f927 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020303e409824fd5916368bdcfd9202e : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89fcf70b3d334c548d3f8de8034e047b : SUCCESS in 6m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c6dd167353441fbb524b7a1d5d6a044 : SUCCESS in 1h 01m 19s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0a9427cc43f841bf93c5c82421813ccd : SUCCESS in 1h 21m 47s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a54622b5baf14d4681365da5d8509712 : FAILURE in 1h 27m 24s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/d84ca46771874f9681b22a38f38c97da : FAILURE in 40m 04s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/1e43c0bea82948b0a4571058c72dd0c2 : SUCCESS in 20m 46s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/6779296d27ce4379a5b93ac1701ce821 : SUCCESS in 31m 27s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/b69cb747ba11403ab3964650988f2226 : SUCCESS in 18m 38s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/80f311e30a414218a479fdc5356b1c74 : SUCCESS in 30m 59s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cdf1419b8e6a4b3a88b5978919e33827 : SUCCESS in 44m 45s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0576e537fdfc47459b7fb52952a78474 : SUCCESS in 1h 08m 27s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/3e8104aaa54e48cda915cd369c2a6fb4 : SUCCESS in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba598f66e681a6e6663a20c8e2b90e2db3d526d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 21:09:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68d0ec4b2d14bf1d1debfadc88d723d260a1947c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:10:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e986907649c8aeb5e2e7bce1300c026d2f0f50a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:27:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9beeb0dcd2f64eda98d9d74bc78c19a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/568720f6f05e40fab26ca64fc69edb16 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7dcc904ed0e4dcc9cc258afe66e56ad : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/000ff1becf9643888741e3059183cca0 : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ca253c479abd4e34a3c7eabaa36bffd0 : SUCCESS in 8m 18s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/17c8ff7ac63b49e9b6811a721feb9a16 : SUCCESS in 35m 14s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4164c92bac2f44e894dcd6bd9bd8db6e : SUCCESS in 4m 04s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/07409cd162e24a3489d2b4b21788b063 : SUCCESS in 6m 30s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b8ff13cd92834aea9697b1ce21aa5455 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f679e87a4a5d47879136eb3f346616a4 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea2b5cbba904322ba9b3878111f5626 : SUCCESS in 4m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/094df8ef2d9341f8a19a18211b5393b8 : SUCCESS in 1h 01m 37s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/8ee8e4d5a8f248a9a16eb25eb9dc101a : SUCCESS in 1h 04m 36s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/9f1dae2bea8849cf9b20f4a745063f86 : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9a4741a9e1cbd2319e8c8ac41a45174915c89651","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:27:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ebab9445d072b7f87bbfc302390f12b661eb5375","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:29:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ae304959e64bdaa2cd53e35b5d1c78\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76239735f6d04506ab45ee87db4e2aa4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af74f7bd3cb3366517dbbac45625112feccf5758","revisions":{"af74f7bd3cb3366517dbbac45625112feccf5758":{"kind":"REWORK","_number":1,"created":"2026-09-07 14:17:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/29/1004429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/29/1004429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/29/1004429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/29/1004429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/29/1004429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/29/1004429/1"}}},"commit":{"parents":[{"commit":"f6fb581d3d46185eaaa973478f45146387b1dc31","subject":"typing: Simplify QueryParamters type handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f6fb581d3d46185eaaa973478f45146387b1dc31"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-04 15:34:20.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-07 14:16:50.000000000","tz":60},"subject":"typing: Simplify QueryParamters type handling further","message":"typing: Simplify QueryParamters type handling further\n\nWe only support CSV values. A more generic type is overkill until we\nactually need it.\n\nChange-Id: I0933e7ea5834d4390c15b7aa9dd4bc9698ada127\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/af74f7bd3cb3366517dbbac45625112feccf5758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/af74f7bd3cb3366517dbbac45625112feccf5758"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
