)]}'
{"id":"openstack%2Fopenstacksdk~1004797","triplet_id":"openstack%2Fopenstacksdk~master~Id9e8c0812edd9bc7edd188ec6280e2a5fe70a300","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-09-14 23:07:44.000000000","reason":"Change was submitted"},"39399":{"account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"last_update":"2026-09-14 23:07:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9e8c0812edd9bc7edd188ec6280e2a5fe70a300","subject":"resource: Fix re-entrant alias resolution","status":"MERGED","created":"2026-09-09 10:26:09.000000000","updated":"2026-09-14 23:10:33.000000000","submitted":"2026-09-14 23:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004797","meta_rev_id":"40014188b4353e9dc03bc47a3d2ae9b943530a02","_number":1004797,"virtual_id_number":1004797,"owner":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 23:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-09 10:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 21:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-09 10:34:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-09 11:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 09:47:14.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"a42d7429d90a47d2de1abd33144947aaf24f86b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-09 10:26:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e274e75b0dc56c40225f88c423cbc3584a6c3424","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-09 10:34:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8789e524d1bebbf0acda07c58bd93f8a50dc4fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 11:54:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48e6c39c20f4dc789cf34e4041324c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29c680da81b540639b72817f35bc62bd : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fdf06d7cb49c4f30ab8ce0fff33f33e2 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32db9cfa9fe64e6fa63cea695382148f : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7c13c7d4058429c8dce304674aa99b8 : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/abc0c4ac6d50406d9fd95dde5e18034b : SUCCESS in 7m 32s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3447279dd341403b849cf383d777e1ee : SUCCESS in 7m 05s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6b3c3db334174efea1d11ccf889f1a9b : SUCCESS in 28m 04s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/67deb98a2e0240498a08f62ef744808e : SUCCESS in 4m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/65e3e3b7b8cb46dc84e718f550e7f5fb : SUCCESS in 2m 50s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/eece9c3f45564657ab48a36fea4524f6 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b649515eb24cb791c85c15c6b68f92 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61643f283f45400bac3eeed18827dc48 : SUCCESS in 5m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2d332dd4d934985a043a6154055f9c9 : SUCCESS in 1h 11m 16s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2ff3ce039f3f445a81c291e5b5691d18 : SUCCESS in 1h 11m 33s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/40c6755ac75c49f6b8823c7759fc6591 : FAILURE in 1h 03m 22s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/e5f1273ea16c43049ce85325d671fe03 : FAILURE in 45m 03s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/51704d123d8e4ea69944759024e482fe : SUCCESS in 19m 11s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/3b8d73ca10e7431d8ce71f56f87f9af2 : SUCCESS in 23m 22s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/27e4ac410bdc47d1a37c009682b71891 : SUCCESS in 31m 04s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/3b9372c78848458c9ec84d5211f9813a : SUCCESS in 25m 22s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/cf9af65279d14393aa97e3c952d585a5 : SUCCESS in 44m 18s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ebc995d08327455f91ff2631e3ebba00 : SUCCESS in 1h 20m 15s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/0dea30c345e04c1a90d57cb4853ab570 : SUCCESS in 1h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2374b9b10009a701e95bb24be235f21bcb98a633","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 21:46:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ac28a8dc41a741b118ac12c0382f322cf022925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:47:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"499d6543edf12220c8aeedc77e86e66c3d69c0de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 23:07:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aaacb03b76d45fb9cd59e975592094e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53bb77e756b34aba935d33afe82bbaa7 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bf5363da2e645fda8a8fd5f69ca2fdb : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44bd554885614b1f99422303bc4d1e72 : SUCCESS in 4m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3acf614b5ffc4108b9a5ecd1cb806249 : SUCCESS in 8m 27s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7fa04857f8634b8d9ed4cd6c9d6cf1e2 : SUCCESS in 1h 00m 35s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/be20ccdb830b47bb844418736b9201de : SUCCESS in 5m 48s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c879ea7d99a64c268cc5125c22c99d86 : SUCCESS in 3m 25s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b9b4e97f45b54687a3be497ebe631d2e : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bda375231113452d938daeda32f5e902 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad171e12d8c644948f9e4d693f3d5e37 : SUCCESS in 2m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19fcd262a86046b4870d64e10ffcaa92 : SUCCESS in 56m 33s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9d0f8d04b78d4d87a8e5fd4dcfdbf853 : SUCCESS in 50m 56s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/7d995d2afd024435a6d90390873667e9 : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":1},{"id":"199b6c698cfd57b1e3676fe962b2f64105f2ace4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 23:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40014188b4353e9dc03bc47a3d2ae9b943530a02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 23:10:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d125f2b14ee4ca2a10d70285bf679cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6198851aebea4093a33293952855efb9 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55def8e1e4527bc09e6bcb870091e95de12900b4","revisions":{"55def8e1e4527bc09e6bcb870091e95de12900b4":{"kind":"REWORK","_number":1,"created":"2026-09-09 10:26:09.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/97/1004797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/97/1004797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/97/1004797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/97/1004797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/97/1004797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/97/1004797/1"}}},"commit":{"parents":[{"commit":"76829ff27484f6a62f8ebe9f9b147ca5a4ee4dd7","subject":"Merge \"identity: Remove TagMixin from Domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76829ff27484f6a62f8ebe9f9b147ca5a4ee4dd7"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-09 10:09:25.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-09 10:25:24.000000000","tz":420},"subject":"resource: Fix re-entrant alias resolution","message":"resource: Fix re-entrant alias resolution\n\nPrevent nested alias lookups from removing the `_seen`\nflag of the outer lookup.\n\nCloses-Bug: #2046838\nChange-Id: Id9e8c0812edd9bc7edd188ec6280e2a5fe70a300\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/55def8e1e4527bc09e6bcb870091e95de12900b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/55def8e1e4527bc09e6bcb870091e95de12900b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
