)]}'
{"id":"openstack%2Fopenstacksdk~1004994","triplet_id":"openstack%2Fopenstacksdk~master~Ie3a18edf66a79fbead3642d827fc2359bdfdc02a","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"39399":{"account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"last_update":"2026-09-16 07:46:13.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ie3a18edf66a79fbead3642d827fc2359bdfdc02a","subject":"placement: Add support for resource provider allocations","status":"NEW","created":"2026-09-10 11:43:35.000000000","updated":"2026-09-16 07:46:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"4b87fbd4ea9a0b64495bf74748d854ef7c855fb4","_number":1004994,"virtual_id_number":1004994,"owner":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-11 04:47:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2026-09-14 21:46:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-10 13:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-14 21:46:28.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"574b83976426d4c4e23b76466fc3c3c3af660351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-10 11:43:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf838469de3610d0e1289679030fe1c332431f28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:19:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab821e3369904320b432779962b7b1dc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5affd73c954e6da58a38e54feb4b58 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb92aaf2ed724f2eaa3a8faa0b21a3e7 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c70f97d029d14c9a83ab1d387a593a53 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20705cba54da43a181ca34177bfaef47 : SUCCESS in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ae946364171045e7a20df72cc3e3a32c : SUCCESS in 6m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b9d1a096b77047de8f550e4dcdc77108 : SUCCESS in 7m 58s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cbbf1c0eca504e5f80345954ece8c633 : SUCCESS in 1h 01m 44s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/81e8eb51a0844306b7382938338dd8ca : SUCCESS in 4m 09s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/31844c10ecf849fabed7d16388948951 : SUCCESS in 6m 01s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/36c8ca73857c49f783e6de303ca8b63f : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67b31b120464811890b6dde7d434a00 : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a5e4fc1ff9d4b5ca124c45af05a7f66 : FAILURE in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fc2ccf54fb84a129e715b0486710567 : SUCCESS in 57m 18s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/91a7c5df46e7401fad583cb9098c4885 : SUCCESS in 1h 24m 15s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ba4786cdd19b4061973ac637a2b886df : FAILURE in 1h 29m 00s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/b199c7e30ee5472daebdd16a52db750d : FAILURE in 54m 55s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d3af4609f3f540b187087d372a6bf591 : SUCCESS in 36m 25s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/80d1973fc6da40eca9a62e9a673f3c82 : SUCCESS in 30m 58s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/e81dadc415664a81851c51d1366b588c : SUCCESS in 32m 26s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/052e243f9359410aa68eee9a2e7f9692 : SUCCESS in 31m 11s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/a2117a08aadd4ccfa0e42a935dcf5ad2 : SUCCESS in 44m 21s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ae64ab1fe8324b3f97d45748645b53b5 : SUCCESS in 1h 10m 24s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/bbf193c4c6f54979b56411564ad6478d : SUCCESS in 1h 25m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86905f1bff2991fd0902e52b88825df736a84eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-11 03:11:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5be75692e408aa7417b945ebb3c8ca1c89ee2341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-11 04:47:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e1bd675ecca499591db51f25a290aee\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35352482b314ffcae7ba17b7b87d5df : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f18790e98e8491dab826a935bb839d3 : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/53519f0e1fde43b5a151542a11896ef3 : SUCCESS in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c43f1b3d806b4d649aca032d489746c1 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/552c0fb4e8894dd8a0d458fb6b9cb946 : SUCCESS in 8m 41s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/94a8c4e5e98f427eb97f7c2503c82834 : SUCCESS in 6m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dfdefd0bf42f4b8183d4d0e9fe505c78 : SUCCESS in 28m 47s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a1818aa01fdd472bbce4bc0994e0f674 : SUCCESS in 5m 15s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/199e0ed9aaf54536ac197c33cacbb816 : SUCCESS in 7m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/69bcdb40f6e146cfb6385b081f6b0aeb : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4440a83705ac46078724e9a21b4cce94 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f10c632c8a9843539ba5fb1969e475c8 : SUCCESS in 6m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f26cf9032df74c4a85ea64205279051a : SUCCESS in 57m 23s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/bbe69725400e49e8b6adc9de8d37d118 : SUCCESS in 1h 28m 40s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/a4b2908fceb54feea61ba24b75036ea6 : FAILURE in 1h 25m 03s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/58c5a017036f46a093a4d7f7cac9305c : FAILURE in 1h 00m 51s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/d030d5164f58488c848a80ec91130aed : SUCCESS in 39m 25s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f4bcea1c6ed149e98901f1ce753cdd18 : SUCCESS in 30m 21s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/c64d4665cb8f4378b58a03a00b86cd09 : SUCCESS in 29m 07s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/ec65024a652d465d9f298c69556a9c9c : SUCCESS in 21m 14s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/fa2b6ec92c7d424faff021989e2ffb9c : SUCCESS in 43m 17s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7f633ef8d87448b8a53a0a1581c05649 : SUCCESS in 1h 08m 48s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/f908bf02a4c94ec08db7cbc19ce46b05 : SUCCESS in 1h 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"66018b150c97540c9302b36a3a367719f4b5e6e8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 21:46:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b87fbd4ea9a0b64495bf74748d854ef7c855fb4","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-16 07:46:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a415e92937785f87cd6d58ece5fc038cd08d04f","revisions":{"c76b3763bb3aaa2810bb75038457d575baf09bbf":{"kind":"REWORK","_number":1,"created":"2026-09-10 11:43:35.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/94/1004994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/94/1004994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/1"}}},"commit":{"parents":[{"commit":"76829ff27484f6a62f8ebe9f9b147ca5a4ee4dd7","subject":"Merge \"identity: Remove TagMixin from Domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76829ff27484f6a62f8ebe9f9b147ca5a4ee4dd7"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-10 11:39:35.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-10 11:39:35.000000000","tz":420},"subject":"placement: Add support for resource provider allocations","message":"placement: Add support for resource provider allocations\n\nAdd resource and proxy methods for retrieving allocations associated\nwith a resource provider.\n\nCloses-Bug: #2084231\nChange-Id: Ie3a18edf66a79fbead3642d827fc2359bdfdc02a\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c76b3763bb3aaa2810bb75038457d575baf09bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/c76b3763bb3aaa2810bb75038457d575baf09bbf"}]},"branch":"refs/heads/master"},"8a415e92937785f87cd6d58ece5fc038cd08d04f":{"kind":"REWORK","_number":2,"created":"2026-09-11 03:11:28.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/94/1004994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/94/1004994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/94/1004994/2"}}},"commit":{"parents":[{"commit":"76829ff27484f6a62f8ebe9f9b147ca5a4ee4dd7","subject":"Merge \"identity: Remove TagMixin from Domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/76829ff27484f6a62f8ebe9f9b147ca5a4ee4dd7"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-10 11:39:35.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-11 03:11:13.000000000","tz":420},"subject":"placement: Add support for resource provider allocations","message":"placement: Add support for resource provider allocations\n\nAdd resource and proxy methods for retrieving allocations associated\nwith a resource provider.\n\nCloses-Bug: #2084231\nChange-Id: Ie3a18edf66a79fbead3642d827fc2359bdfdc02a\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8a415e92937785f87cd6d58ece5fc038cd08d04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/8a415e92937785f87cd6d58ece5fc038cd08d04f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY"},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
