)]}'
{"id":"openstack%2Fopenstacksdk~1005380","triplet_id":"openstack%2Fopenstacksdk~master~I75a642b3ff831ea6ff37ee15ebcaeb19c165c12f","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"I75a642b3ff831ea6ff37ee15ebcaeb19c165c12f","subject":"functional: Reorganise cloud functional tests by service","status":"MERGED","created":"2026-09-12 12:02:25.000000000","updated":"2026-09-14 22:52:25.000000000","submitted":"2026-09-14 22:50:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005380","meta_rev_id":"ceda550617c088f4b83d690d2484abf0b14756ea","_number":1005380,"virtual_id_number":1005380,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 22:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-14 21:10:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-14 21:10:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-12 13:40:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"861ab8b07e71e545cdeec4f23e357fbce5ecfb1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-12 12:02:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09ee85648d754a3381689fbfb3c7c9b8ca951431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-12 13:40:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752081a1702f4b93999cf65792b8c3bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c1ebf54fcea44c38d1c01a3f844db58 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a35e222d72cb455fb5edb0e5fc85536d : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e199d5f7d824ebb807b2095602a98fb : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da203a1f5198461182efbc9ceac455e9 : SUCCESS in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45c4c7f2e8da43f9891c66ff7835c59b : SUCCESS in 6m 46s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/83dce9515e4c4779bed8d7fc62f754b8 : SUCCESS in 4m 07s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5cec2feacb9941c2bb599b29e9348b27 : SUCCESS in 59m 12s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/48cd78feb6a84666a271505c3ed1af5d : SUCCESS in 4m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f0eb59b621384d0bb612a03b9d9b2553 : SUCCESS in 4m 54s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/5d580ceb426d48fda4a3128f17cdc64e : SUCCESS in 7m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc45dc22bc874e368e5510c58c18a60d : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5c434b2b1d3449e94d074b42ca8f626 : SUCCESS in 5m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c6d1ea13232470f97c2cc1cd4206247 : SUCCESS in 31m 27s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/921d1c072311469ab6ed8fafe5dc2c50 : SUCCESS in 1h 06m 39s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ec591cfe926c4c4792216f8c285aac13 : FAILURE in 1h 34m 09s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/209f42e9760941da8d42eec2893f855e : FAILURE in 53m 26s (non-voting)\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/468da105f487430f9566bb1963e995fe : SUCCESS in 33m 29s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/bede64b5741b4d95b0a48f4c64664f51 : SUCCESS in 15m 21s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/c7ae057c130d4960bb2b0d854ab61a3d : SUCCESS in 26m 13s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/878e5d3719aa45808f20b15e162e19ee : SUCCESS in 31m 10s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e6cfd957947a488283266b97f6aa5021 : SUCCESS in 30m 03s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ae6800ea8ba74d83b32ef982b14528b0 : SUCCESS in 42m 53s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/2f9a990dc1064b53a4c3a52c444054b7 : SUCCESS in 1h 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1acfbcb68018de38742f1f59f71fec70b917cdc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 21:10:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9915c73b34d3fb6c205b35bee1603034d69f793a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 21:10:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f2be2e101625e676bc4818287f284488752809f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:50:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1122325f521044908a9cb0abc2ea6294\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ecf3e867504424bb3ee98d00094ba1b : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f2077ca66294f8384215c478e3e78fd : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7687c26bff2147a990b467fb97248460 : SUCCESS in 3m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1771b518f5254d0eb68c705b489bdde7 : SUCCESS in 6m 54s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/599c160c8da84b8ab34f17f5634d2517 : SUCCESS in 29m 07s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9ec0fdf8c5484edd9e0b6b9cbe5dda0b : SUCCESS in 3m 02s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e871f562097a42f999d920d7dd53ec1a : SUCCESS in 6m 19s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/97e1dbd29de94db4b8cd1f35dee3b902 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70bd74c1a9934ff0920f08ad7044d1ab : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cfd5594ab644c6c8b79c6f711f980d0 : SUCCESS in 4m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f751d523cef46cc8955236d31549008 : SUCCESS in 33m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/0a01af80c5154447a993df19a4bbf760 : SUCCESS in 1h 05m 59s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/42722c4527f244a5971e459f091693a5 : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b3d6759f1e9d4bbece17fc4a2408ed2ced8a4ce9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:50:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ceda550617c088f4b83d690d2484abf0b14756ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:52:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd1dedfba3a4bb784663c29a054aff7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d30bd4925546b0b7abae8855588d68 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7189458a3fe49d8141eabeaf73ceda3e48853f23","revisions":{"7189458a3fe49d8141eabeaf73ceda3e48853f23":{"kind":"REWORK","_number":1,"created":"2026-09-12 12:02:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/1005380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/80/1005380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/80/1005380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/80/1005380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/80/1005380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/80/1005380/1"}}},"commit":{"parents":[{"commit":"45e4c43181c4778a60ebba88043c67ef9c71945f","subject":"key-manager: Fix SecretACL proxy and broken functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/45e4c43181c4778a60ebba88043c67ef9c71945f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-12 11:44:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-12 11:49:47.000000000","tz":60},"subject":"functional: Reorganise cloud functional tests by service","message":"functional: Reorganise cloud functional tests by service\n\nRename test files in openstack/tests/functional/cloud/ to match\nthe module names in openstack/cloud/. This means each file covers one\nservice rather than one resource type. Doing this should make opting\ninto tests an easier feat.\n\nChange-Id: I75a642b3ff831ea6ff37ee15ebcaeb19c165c12f\nCo-Authored-By: Claude Sonnet 4.6 (1M context) \u003cnoreply@anthropic.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7189458a3fe49d8141eabeaf73ceda3e48853f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7189458a3fe49d8141eabeaf73ceda3e48853f23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
