)]}'
{"id":"openstack%2Fopenstacksdk~1005692","triplet_id":"openstack%2Fopenstacksdk~master~Ia0b892f782179a099c5f48f51620484cc0705dc0","project":"openstack/openstacksdk","branch":"master","hashtags":[],"change_id":"Ia0b892f782179a099c5f48f51620484cc0705dc0","subject":"typing: Remove unnecessary cast","status":"MERGED","created":"2026-09-15 09:39:19.000000000","updated":"2026-09-16 13:58:03.000000000","submitted":"2026-09-16 13:43:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005692","meta_rev_id":"38ef485928e6b6599ac1a1cb4b2dc0cd2edca95c","_number":1005692,"virtual_id_number":1005692,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 13:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-16 12:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-16 12:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 10:53:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62489810f664a91df1f443f9b1db690d9cac1c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 09:39:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"690efd4c4e3861ac6ed83f4200f12feec4713980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:53:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/623b10767c364c168348cfb9f94e4fe6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c554ff5cddb44f0486e4839eacef9bea : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a24d2173a96c48b0acd68d16fe38a786 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20730359323440b9aa4364723607cdcc : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cb488adb0a546d687f4cd3640921371 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a7cbd780de584d4b9398ec0bcda938b6 : SUCCESS in 4m 26s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/a6ff7a5a3fc54804bb5ca5460744d7f3 : SUCCESS in 8m 52s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6980b1ff67964893adef90f3cc006679 : SUCCESS in 40m 58s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b65612b9500a48988a466355c423c6dc : SUCCESS in 5m 20s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ff8350e9a580469cb886f0058470da41 : SUCCESS in 5m 05s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1c884c5237b64f0db7e0daa421250318 : SUCCESS in 8m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2069dcbc564b45b6afb7e5efe46a14da : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f86431b3ff841ae98b47c312b75dae3 : SUCCESS in 5m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0866958d445447179943721e75e042fc : SUCCESS in 40m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/53ec8cbc281648fe91a0b82fd3b333f5 : SUCCESS in 1h 10m 55s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ff9e9681294b46e3b03bbfd430d2f83f : FAILURE in 55m 07s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/9d4ad22059f5415488fc80f62fed662b : SUCCESS in 21m 24s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/04c8eda4a8de452e8766dd88ed609c72 : SUCCESS in 34m 40s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/fbd0834752114f49a10362cb613ec9d6 : SUCCESS in 28m 50s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/40754f71d72c4fde848b8dfd5360a688 : SUCCESS in 25m 54s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/bc750334f0fa41a9a783f8fb290c03f2 : SUCCESS in 30m 01s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/dd5821e11b0e4d28a003376f06998304 : SUCCESS in 27m 29s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1fa6b49dc16e4986ade0b1a035b00dff : SUCCESS in 1h 07m 40s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/745f594150394b188f4bddb36476b450 : SUCCESS in 39m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7e028d52af13417981d6e34f2585e6ba2b259c3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 12:11:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d5a7667b142f1ada75afa0ac7a418f974738d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 12:12:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44bab889d7e8249187f9acc3f0e6f9dbbd0933c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:43:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a68a156f0704a50a2c3140616aaf58e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/190048645a784f42897d52c51352f7ed : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de88f9a2cd1a4fcabf1491b39c22e955 : SUCCESS in 6m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bc39fe6749c401d89a437a2c096205f : SUCCESS in 4m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1788c1ebc51841ac956084629aa3d7df : SUCCESS in 7m 17s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c235bc2bda1e4dc196feff54af0adfcd : SUCCESS in 1h 02m 00s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a2d58321bf0b44de921525b54641fc1c : SUCCESS in 3m 33s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c8e998a076e64b46a1540743f03458c9 : SUCCESS in 7m 42s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9ca32ea2c00d4bbc85b513de256d464c : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3946066e2a844bf29e4e7365a0b3defe : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5e5404ea71d47f8b67013bc9a3b190e : SUCCESS in 2m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/318a7242431f448586c1a39f09d308bc : SUCCESS in 30m 23s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/9f7439e11a384a2c8b1664b3b9046928 : SUCCESS in 1h 28m 01s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/5ba241298e5f417db027f4a84301c810 : SUCCESS in 32m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3c4aa20a86aa3c00662d4b17ad13ca876d9d55e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:43:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"38ef485928e6b6599ac1a1cb4b2dc0cd2edca95c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 13:58:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05d26856f1cc42b3bef1d3e4362d297e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747370def1604c47801f57a02d0c30f8 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","revisions":{"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56":{"kind":"REWORK","_number":1,"created":"2026-09-15 09:39:19.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/92/1005692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/92/1005692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/92/1005692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/92/1005692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/92/1005692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/92/1005692/1"}}},"commit":{"parents":[{"commit":"6f646ce6e4e1d336d609d3fcc2e97eae87174d83","subject":"Merge \"resource: Fix re-entrant alias resolution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f646ce6e4e1d336d609d3fcc2e97eae87174d83"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:22:41.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:22:41.000000000","tz":60},"subject":"typing: Remove unnecessary cast","message":"typing: Remove unnecessary cast\n\nChange-Id: Ia0b892f782179a099c5f48f51620484cc0705dc0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
