)]}'
{"id":"openstack%2Fopenstacksdk~1005733","triplet_id":"openstack%2Fopenstacksdk~master~Ic55571e93dc8aaabaaced67628b4053ba1716225","project":"openstack/openstacksdk","branch":"master","attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-16 14:47:13.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-15 14:36:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"39399":{"account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"last_update":"2026-09-23 11:48:25.000000000","reason":"\u003cGERRIT_ACCOUNT_39399\u003e replied on the change","reason_account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"}}},"hashtags":[],"change_id":"Ic55571e93dc8aaabaaced67628b4053ba1716225","subject":"resource: Fix CSV query parameter formatting for strings","status":"NEW","created":"2026-09-15 14:14:34.000000000","updated":"2026-09-23 13:21:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ac0cf142fee476335a1b4284a2b4c4a9038ffd3","_number":1005733,"virtual_id_number":1005733,"owner":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-23 13:21:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 14:15:12.000000000","updated_by":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-15 14:36:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-09-15 15:46:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d6730bb8f17c7a4a56535996bf00933a7e4b989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-15 14:14:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af1717613fbcfb6b23355b185650c144b237988d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 14:36:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7ce70cac4a67a9c8865d37b74fb0f0ed98c65a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 15:46:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588f0f941f3440799727198111474b00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e86b651a72c4ab495ca0ae2b31425f0 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75082afca33342d5a3a7addffebcdee1 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8f1cc69a53a3412ebb4a36d072349e1f : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ea1504132694b408c4f1344f8fe4d1f : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/881bc2a8bab649eb9d83023ea775bd45 : SUCCESS in 6m 47s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/04aa63a2824448149183b3d317af37e2 : SUCCESS in 6m 56s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/22792243eca443ce9f81468db3cbef74 : SUCCESS in 46m 29s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/69f3d8f8f468400186dcebf8765336ed : SUCCESS in 2m 46s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a807c8abfed04eef8670ff6e638ba616 : SUCCESS in 3m 01s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/9dadbe0b652349a9a5da7cca2f1c2615 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22142c7090984c7a9d2a1c16769efc59 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f88127cc4d854225abd24f2219d17c84 : SUCCESS in 4m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d00bfb9744ed4b78a98a62ab8d15a886 : SUCCESS in 1h 00m 36s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c2da0181b605482292bef3d8060b0727 : SUCCESS in 1h 26m 00s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/6f34592babcb4a03978c5ac9a8fe54ca : FAILURE in 1h 09m 20s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a303b9d745734426919942f7e6e948c1 : SUCCESS in 43m 08s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/e61153a39d064fc18349777997b384b9 : SUCCESS in 27m 12s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/f423feb750bc47049a5b1916ea406bfa : SUCCESS in 28m 18s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/0f8c67a2c415466f812df50224130270 : SUCCESS in 19m 33s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/5fe8aafe6f314e62a2ec02dd0070456d : SUCCESS in 21m 47s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/1523e41dabaf4e40b2912efec1b91afb : SUCCESS in 30m 16s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0aa667dccacb4b3dbfe397a4d0956d24 : SUCCESS in 1h 13m 04s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/671ac61025ce4644a754ee8ac59c46d5 : SUCCESS in 1h 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02f0df3d81355505aefd94e33cd9228402d0283d","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-16 07:53:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9e39ba54c9c7e318993de1d1069030f4c120be52","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 12:12:36.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"642b01c2c3f37e04638e9ceeab4930b797ca5afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-16 14:47:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c7d3b2a30e1d5d4656f0c885cf086dc67d1723bd","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-16 14:47:49.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"84ee7927055e40f4fff279cc7c3831aea4d2cfc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:20:03.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/083649bfbf364cbea185138d02066a7e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87a7619d8794c808846d32adc598cad : FAILURE in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/739990fed84e455e8e42239daaaab962 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60fec29014c8446b8c9f2aac9d77eb98 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd3af25e367643b28c644afeb88776a3 : SUCCESS in 3m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9bf732f785a94b549ad98c6e4ee5c55d : SUCCESS in 4m 12s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/20082b90610f496eb0f851d35dacd1b6 : SUCCESS in 6m 35s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0a0b8b38273747699a9fc4b76c6bcbce : SUCCESS in 1h 05m 08s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e2b7455c4d1849ea9ae3fc157632661b : SUCCESS in 6m 06s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ac52578e67a647348a17eeeb107f3385 : SUCCESS in 7m 07s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/2e964be753c7403c80c2bbfd2d446806 : SUCCESS in 10m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d6a049bf884984b0ec482f780d02f9 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ef4f52c9b0a4312bb35a60524ba2fcd : SUCCESS in 4m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/653baa72c76b47dc9b9d9fcb8295d85a : SUCCESS in 1h 01m 55s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/930f05472d154a0994d18bf0b1a5f322 : SUCCESS in 1h 27m 21s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/f2deef1b6b10437f9a47bc3a57d3df22 : FAILURE in 54m 06s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/a1d29ec9a03a42c0b3f5d8141c38dc2c : SUCCESS in 46m 57s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/4f8de8a4181543d58be55f9696439da9 : SUCCESS in 38m 44s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/751d2b60e9bd446ca775b57ab55d1c67 : SUCCESS in 17m 41s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/9072cf4c1b454e87bd1c60f8dad611bd : SUCCESS in 27m 08s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/21654c1d09e742ddada5ef123a3c68ba : SUCCESS in 14m 59s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/f03ec71a591549df810ee8ea1e71d1d5 : SUCCESS in 18m 59s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c8c92c5a361e4ee6887bb669315bacfc : SUCCESS in 1h 09m 28s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/620d7036c3884d82ae05b05000f4e004 : SUCCESS in 1h 06m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd30f13bcef0f2046ac13c443457f9924a86d0ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-16 16:34:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ac39bba220035c3dd9432a70893aa9b80237a75e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 18:07:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d9c5965baa4ece89d303685f6ac23b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d80553898b4fa1a61e071ffe3d8991 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfb961face4449d382554da8c8f50800 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ee7c0b714414ad8902e60cdec273c0a : SUCCESS in 4m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f27f2013790b4584ab4f7acb93af25bd : SUCCESS in 4m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e862ba320b8e4de2a28ff808139d8dd9 : SUCCESS in 4m 14s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d4936a2880184044bf434c2b8f5680cf : SUCCESS in 6m 38s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a5e9f11d4e384511bec077f87ca269fc : SUCCESS in 46m 13s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/289b9656a67f4259b55627d43efcc005 : SUCCESS in 2m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2d64c1034e3c4b57ac3930c9522dfaaf : SUCCESS in 5m 03s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/b467089de2704063a6643a786978576d : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c56f35c3714ef8947ed16b47005f86 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ea22e1c111f4ecd8e48f790d3dce43f : SUCCESS in 3m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/156056f074714a05b56f9d3af24738ea : SUCCESS in 1h 05m 14s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/c446a818eed74a93b60bafff9632e82f : SUCCESS in 1h 24m 45s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/5fcffe6211e84c28a5201fb4d15ac73b : FAILURE in 1h 25m 24s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/ec8281bbb2ef45509674b0fe83b5f1fc : SUCCESS in 42m 27s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/198bc2a5cd55462cbbabf93a5faad03c : SUCCESS in 34m 27s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/50181f2fc3454b428f14a1caabc9567b : SUCCESS in 30m 29s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/74b6b80b58d548a39ea06ca6b58c6677 : SUCCESS in 30m 26s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/2dd0002cd1cf48edb5e3b8e3b75e01f6 : SUCCESS in 33m 12s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/500e70a670984563b884fddf0f494402 : SUCCESS in 39m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0acf41b2b8cd440cafeca21049d5537f : SUCCESS in 1h 12m 31s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/54ff57d4d5dd40b4b3e914f6e7c54753 : FAILURE in 16m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a130bafae10df69ed4e0340f5005fbf7062c67f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-23 11:47:24.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1b984fb3a6c3b4b12b0344743a537816da9f0b26","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-23 11:48:25.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2ac0cf142fee476335a1b4284a2b4c4a9038ffd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 13:21:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8589a551c3d64b26b29cbfded3c2c358\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d05ed1a059c480a9feda08c6af68c63 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5bc58403b5146bb8183319c20f6c576 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fec418ec6dcb48f8be42f52eace9e6c8 : SUCCESS in 4m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f998f885e52e49ddac0eb37deeb17e84 : SUCCESS in 3m 51s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/2f5d4086223042a6bc49fca9393a6f1b : SUCCESS in 8m 05s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ae20a74f76ad468c8ccabb72112e3a62 : SUCCESS in 1h 07m 21s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5dec86a9398248298b4bbcea2efa5e02 : SUCCESS in 3m 49s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ba4efb12d4ad46a29f184afe78da6356 : SUCCESS in 2m 36s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/af765d1fcd0c4fc1a4dcfa3c650d1625 : SUCCESS in 9m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951cabf4572d4aac9f8ac9021d358e6f : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a33ba137457e4600bdf139f2168122af : SUCCESS in 5m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cbdddefeda746b8be43342abbd466ec : SUCCESS in 54m 54s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/598288dc7978441db8a5815877898bd9 : SUCCESS in 1h 24m 29s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ab3a047b0fd0459ea06c0cfcad9ecfc3 : SUCCESS in 1h 27m 15s\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/72f4792c263f4bfabb5ba56cc7b49fa9 : SUCCESS in 19m 08s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/abb4331fa0a44d74898d74a710f31e59 : SUCCESS in 36m 16s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/2b78dfb32c4745058cb88df8a33e7738 : SUCCESS in 29m 47s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/51376993f769462a9f4523ceebbddf99 : SUCCESS in 28m 13s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/216cd7df438e4826a5e3a8f0a013010f : SUCCESS in 30m 04s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/ca9a80fa835e46dbb9000bada732e86c : SUCCESS in 50m 32s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d16bd18744e5405aab5d48f70c8032e6 : SUCCESS in 1h 14m 17s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/ac5d2d67867f470da3d0abc27e49a92a : FAILURE in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7488ad1e0f5a689d66f5a846c61a394777273c69","revisions":{"123c4ebaebc66f59aa959bfe4b15a65c42a26a44":{"kind":"REWORK","_number":1,"created":"2026-09-15 14:14:34.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/33/1005733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/33/1005733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/1"}}},"commit":{"parents":[{"commit":"6f646ce6e4e1d336d609d3fcc2e97eae87174d83","subject":"Merge \"resource: Fix re-entrant alias resolution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f646ce6e4e1d336d609d3fcc2e97eae87174d83"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-15 14:07:24.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-15 14:11:59.000000000","tz":420},"subject":"resource: Fix CSV query parameter formatting for strings","message":"resource: Fix CSV query parameter formatting for strings\n\nFixed `csv` query parameter formatting for strings and use it for tag filters.\n\nCloses-bug: #2051698\nChange-Id: Ic55571e93dc8aaabaaced67628b4053ba1716225\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/123c4ebaebc66f59aa959bfe4b15a65c42a26a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/123c4ebaebc66f59aa959bfe4b15a65c42a26a44"}]},"branch":"refs/heads/master"},"ca42535ed9f4bddc339c96f1bf816d7269e6d3a9":{"kind":"REWORK","_number":2,"created":"2026-09-16 14:47:13.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/33/1005733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/33/1005733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/2"}}},"commit":{"parents":[{"commit":"6f646ce6e4e1d336d609d3fcc2e97eae87174d83","subject":"Merge \"resource: Fix re-entrant alias resolution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f646ce6e4e1d336d609d3fcc2e97eae87174d83"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-15 14:07:24.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-16 14:46:49.000000000","tz":420},"subject":"resource: Fix CSV query parameter formatting for strings","message":"resource: Fix CSV query parameter formatting for strings\n\nFixed `csv` query parameter formatting for strings and use it for tag filters.\n\nCloses-bug: #2051698\nChange-Id: Ic55571e93dc8aaabaaced67628b4053ba1716225\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ca42535ed9f4bddc339c96f1bf816d7269e6d3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/ca42535ed9f4bddc339c96f1bf816d7269e6d3a9"}]},"branch":"refs/heads/master"},"77edf7a6a0d9ebe544bc5ab8989cb95b2263ec66":{"kind":"REWORK","_number":3,"created":"2026-09-16 16:34:08.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/33/1005733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/33/1005733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/3"}}},"commit":{"parents":[{"commit":"6f646ce6e4e1d336d609d3fcc2e97eae87174d83","subject":"Merge \"resource: Fix re-entrant alias resolution\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/6f646ce6e4e1d336d609d3fcc2e97eae87174d83"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-15 14:07:24.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-16 16:33:29.000000000","tz":420},"subject":"resource: Fix CSV query parameter formatting for strings","message":"resource: Fix CSV query parameter formatting for strings\n\nFixed `csv` query parameter formatting for strings and use it for tag filters.\n\nCloses-bug: #2051698\nChange-Id: Ic55571e93dc8aaabaaced67628b4053ba1716225\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/77edf7a6a0d9ebe544bc5ab8989cb95b2263ec66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/77edf7a6a0d9ebe544bc5ab8989cb95b2263ec66"}]},"branch":"refs/heads/master"},"7488ad1e0f5a689d66f5a846c61a394777273c69":{"kind":"REWORK","_number":4,"created":"2026-09-23 11:47:24.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/33/1005733/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/33/1005733/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/33/1005733/4"}}},"commit":{"parents":[{"commit":"60fcc83c5352080b9147aa299ef6db255bf8357d","subject":"Merge \"placement: Add support for usages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/60fcc83c5352080b9147aa299ef6db255bf8357d"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-15 14:07:24.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-23 10:54:01.000000000","tz":420},"subject":"resource: Fix CSV query parameter formatting for strings","message":"resource: Fix CSV query parameter formatting for strings\n\nFixed `csv` query parameter formatting for strings and use it for tag filters.\n\nCloses-bug: #2051698\nChange-Id: Ic55571e93dc8aaabaaced67628b4053ba1716225\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7488ad1e0f5a689d66f5a846c61a394777273c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/7488ad1e0f5a689d66f5a846c61a394777273c69"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
