)]}'
{"id":"openstack%2Fopenstacksdk~1005901","triplet_id":"openstack%2Fopenstacksdk~master~I54cf0e9eb3aad78b80f3ae400b878fc04994b43b","project":"openstack/openstacksdk","branch":"master","attention_set":{},"removed_from_attention_set":{"39399":{"account":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"last_update":"2026-09-23 04:55:02.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-23 04:55:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54cf0e9eb3aad78b80f3ae400b878fc04994b43b","subject":"heat: Fix stack query tag parameters","status":"MERGED","created":"2026-09-16 14:38:46.000000000","updated":"2026-09-23 04:56:33.000000000","submitted":"2026-09-23 04:55:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005901","meta_rev_id":"7225229e9ebf3a0a7266f4aa142e8e9881634294","_number":1005901,"virtual_id_number":1005901,"owner":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-23 04:55:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-23 02:06:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-16 14:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-23 02:06:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-16 14:40:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-16 14:40:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-09-16 16:27:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-23 02:06:09.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"2d61c8810efabdeb5d074a0ac743d3de1a6bd226","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"date":"2026-09-16 14:38:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"056e8397e8406fe633b9aa6e2e2f78b073ac14d9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-16 14:40:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1bed6975c26f40575d7c60472661f3434d4336bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 16:27:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11ed71b213a44ff6abd1f47fd43c4301\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/131cc38ed1dc46c280bd717a56537016 : SUCCESS in 6m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa0622b87eb1440cb4e811d0c7acc5a4 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0603b0f5555840d1983d90f700547c8e : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f515d894326454cbe365c0cb8db4b2d : SUCCESS in 5m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a20ed58c39f45a895c22d0b2c86e890 : SUCCESS in 6m 54s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/cb2f84fa51b74491a008c023a7896938 : SUCCESS in 9m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d030f001f0c045d7825aaf14f4949815 : SUCCESS in 1h 02m 52s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e3c47dfea226443199b90bb6765fcff3 : SUCCESS in 3m 48s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9178618a02c843629cb5b1e246964d24 : SUCCESS in 4m 55s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/273ae3210bc7449cafb1576c8a56dbe7 : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33a774277d446d2b304de3ec58ea1d4 : SUCCESS in 11m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3dd8241e11f4294b8a700b7da6477b6 : SUCCESS in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b1018953dbe4688bc52a99bd9dcf87c : SUCCESS in 1h 01m 02s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/6f65336bcd014dbc99973e3998b4c31d : SUCCESS in 1h 26m 46s\n- openstacksdk-functional-devstack-networking-ext https://zuul.opendev.org/t/openstack/build/ad3f05c7ad7c42968f559c5fca27ab2c : FAILURE in 1h 27m 49s (non-voting)\n- openstacksdk-functional-devstack-magnum https://zuul.opendev.org/t/openstack/build/2c2fafedb3c74d04af0d2f6ae595838f : SUCCESS in 34m 52s\n- openstacksdk-functional-devstack-manila https://zuul.opendev.org/t/openstack/build/36b5b154923d4fe3a691b3cd2ecc22ed : SUCCESS in 48m 05s\n- openstacksdk-functional-devstack-compute-manila https://zuul.opendev.org/t/openstack/build/3fea9a81bb9643648f77844449a07d21 : SUCCESS in 28m 41s\n- openstacksdk-functional-devstack-cyborg https://zuul.opendev.org/t/openstack/build/ef91ee734bc94682b6871f077c337c2a : SUCCESS in 30m 03s\n- openstacksdk-functional-devstack-masakari https://zuul.opendev.org/t/openstack/build/22dac0e9f4274881a9b322f811ea6394 : SUCCESS in 24m 02s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/faebd25e842a43b7911a01cfcdf38e7b : SUCCESS in 44m 04s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/adadf8c831524c9891598e662d99db7b : SUCCESS in 1h 38m 34s (non-voting)\n- ansible-collections-openstack-functional-devstack https://zuul.opendev.org/t/openstack/build/32019f184ced415b92fe551569bd5359 : SUCCESS in 1h 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2987b3f2ebeab7ae8117ce0e85721517cc526093","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-23 02:06:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6b7923838a864fdd3302bcaee04ed3d2065e707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 02:52:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af8c978418fd2a2b59b7f844ebf8a75e167facb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:55:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c95cc82845434b0d91d9c88b67fe4be7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5e203c6ff944a6a82643fbe733ac7f7 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85c618e40d1444d9931a02197252c6ab : SUCCESS in 4m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ae7b5abed4942609d5bf0c8f849f54f : SUCCESS in 6m 39s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8510a449247344f9a1d7c7db23471db3 : SUCCESS in 59m 38s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e1f0e92188b94d9b9ee9d281d853807c : SUCCESS in 3m 38s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/23cd2131176b41e2974d393b0df932f9 : SUCCESS in 6m 53s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/61249b02761c4d4d9e3f77b070647ffd : SUCCESS in 8m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a059cb573446b0b1a6a766329cff5d : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/912fec6feb674cabafcd8302680f3c3d : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a18beee7d1bf4d9c99fcb77d4be4782c : SUCCESS in 1h 12m 34s\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1880694c34e5427092e24d1d70b2aa04 : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack-ironic https://zuul.opendev.org/t/openstack/build/e1f0cf55efac42d6b2d0afc2e124ba18 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f482c73e2ec5c03d736038eaa9b8e3dfa5ae5095","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:55:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7225229e9ebf3a0a7266f4aa142e8e9881634294","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-23 04:56:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec72a0f5241542dba211b6741ae0450b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18a3acbdb4f41adb30549b269796c75 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f308c8d486edfb4d89ca0092cf7481333bcab45a","revisions":{"f308c8d486edfb4d89ca0092cf7481333bcab45a":{"kind":"REWORK","_number":1,"created":"2026-09-16 14:38:46.000000000","uploader":{"_account_id":39399,"name":"Vladislav Terekhov","display_name":"Vlad","email":"VlSTerekhov@gmail.com","username":"tepex"},"ref":"refs/changes/01/1005901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstacksdk","ref":"refs/changes/01/1005901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/01/1005901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/01/1005901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstacksdk refs/changes/01/1005901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstacksdk refs/changes/01/1005901/1"}}},"commit":{"parents":[{"commit":"b22ad455fc9dcde0e10c125a7d5bc56f515e3e56","subject":"typing: Remove unnecessary cast","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/b22ad455fc9dcde0e10c125a7d5bc56f515e3e56"}]}],"author":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-16 14:32:32.000000000","tz":420},"committer":{"name":"Vladislav Terekhov","email":"VlSTerekhov@gmail.com","date":"2026-09-16 14:36:47.000000000","tz":420},"subject":"heat: Fix stack query tag parameters","message":"heat: Fix stack query tag parameters\n\nStack API uses underscores instead of hyphens in tag\nquery parameters\n\nCloses-bug: #2051698\nChange-Id: I54cf0e9eb3aad78b80f3ae400b878fc04994b43b\nSigned-off-by: Vladislav Terekhov \u003cVlSTerekhov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f308c8d486edfb4d89ca0092cf7481333bcab45a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstacksdk/commit/f308c8d486edfb4d89ca0092cf7481333bcab45a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
